Uh oh!
There was an error while loading. Please reload this page.
Upgrade @primer/behaviors - #1814
Conversation
🦋 Changeset detectedLatest commit: 335827a The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
size-limit report 📦
|
rezrah
commented
Jan 24, 2022
FYI @pksjce I've tested in Next.js ✅ and Remix Run ❌ .. Multiple Remix error still occurring. E.g. |
@primer/behaviors has been updated to output both cjs and esm files now. This needs to be consumed in primer/react
Closes#1762
There are no visual changes
Merge checklist
Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.