Skip to content

ActionMenu2 + DropdownMenu2: Keyboard functionality for wrapping through options - #1828

Merged
siddharthkp merged 4 commits into
mainfrom
siddharth/menu-keyboard-wrap
Feb 1, 2022
Merged

ActionMenu2 + DropdownMenu2: Keyboard functionality for wrapping through options#1828
siddharthkp merged 4 commits into
mainfrom
siddharth/menu-keyboard-wrap

Conversation

@siddharthkp

Copy link
Copy Markdown
Member

Closes # (type the issue number after # if applicable; otherwise remove this line)

Screenshots

Before:
Kapture 2022-01-31 at 14 04 54

After:
Kapture 2022-01-31 at 14 03 48

Spec:

Menu

KeyFunction
Up Arrow
  • ✅ Moves focus to the previous menu item.
  • ⛔️ If focus is on the first menu item, moves focus to the last menu item.
Down Arrow
  • ✅ Moves focus to the next menu item.
  • ⛔️ If focus is on the last menu item, moves focus to the first menu item.

References

@siddharthkp
siddharthkp requested review from a team and rezrahJanuary 31, 2022 13:06
@changeset-bot

changeset-botBot commented Jan 31, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: dc9eb71

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkp
siddharthkp requested review from colebemis and removed request for rezrahJanuary 31, 2022 13:06
@siddharthkpsiddharthkp changed the title ActionMenu2 + DropdownMenu2: Change focusOutBehavior to wrapActionMenu2 + DropdownMenu2: Keyboard functionality for wrapping through optionsJan 31, 2022
@github-actions

github-actionsBot commented Jan 31, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js61.34 KB (0%)
dist/browser.umd.js61.72 KB (0%)

@siddharthkpsiddharthkp added this to the FY22 - Q3 milestone Jan 31, 2022

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🚢

@siddharthkp
siddharthkp merged commit 6a695bd into mainFeb 1, 2022
@siddharthkp
siddharthkp deleted the siddharth/menu-keyboard-wrap branch February 1, 2022 13:22
@primer-cssprimer-css mentioned this pull request Feb 1, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@colebemis