Skip to content

Add package engines - #1840

Merged
colebemis merged 4 commits into
mainfrom
add-package-engines
Feb 4, 2022
Merged

Add package engines#1840
colebemis merged 4 commits into
mainfrom
add-package-engines

Conversation

@jclem

@jclemjclem commented Feb 4, 2022

Copy link
Copy Markdown
Contributor

This adds the following to package.json and docs/package.json:

"engines": {
"node": ">=12",
"npm": ">=7"
}

Node 12 works with this project and since it's the oldest maintenance LTS version of Node.js, it seems like a reasonable choice.

The docs/package-lock.json has also been updated to v2, now that we're using npm>=7, which uses the newer lock file format.

@jclem
jclem requested review from a team and pksjceFebruary 4, 2022 18:49
@changeset-bot

changeset-botBot commented Feb 4, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 6ef39fb

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@jclem
jclemforce-pushed the add-package-engines branch from 58b0d27 to c91c48cCompareFebruary 4, 2022 18:50
@jclem
jclemforce-pushed the add-package-engines branch from c91c48c to 06038b7CompareFebruary 4, 2022 18:51
@github-actions

github-actionsBot commented Feb 4, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js61.64 KB (0%)
dist/browser.umd.js62 KB (0%)

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks, @jclem!

@colebemis
colebemis enabled auto-merge (squash) February 4, 2022 19:00
@colebemis
colebemis merged commit 1c4786c into mainFeb 4, 2022
@colebemis
colebemis deleted the add-package-engines branch February 4, 2022 19:04
@primer-cssprimer-css mentioned this pull request Feb 4, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jclem@colebemis