Skip to content

FormControl follow-ups - #1850

Merged
mperrotti merged 15 commits into
mainfrom
mp/fix-form-control-docs
Feb 9, 2022
Merged

FormControl follow-ups#1850
mperrotti merged 15 commits into
mainfrom
mp/fix-form-control-docs

Conversation

@mperrotti

Copy link
Copy Markdown
Contributor

Closes # (type the issue number after # if applicable; otherwise remove this line)

Screenshots

Please provide before/after screenshots for any visual changes

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@mperrotti
mperrotti requested review from a team and koddssonFebruary 9, 2022 16:42
@changeset-bot

changeset-botBot commented Feb 9, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 319e5ff

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Feb 9, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js62.63 KB (+0.21% 🔺)
dist/browser.umd.js63.01 KB (+0.21% 🔺)

Comment threaddocs/content/FormControl.mdx Outdated
Comment thread.changeset/silly-humans-float.md Outdated
}

const FormControl = ({children, disabled, id: idProp, required}: FormControlProps) => {
const FormControl = ({children, disabled, id: idProp, required, sx}: FormControlProps) => {

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

FormControl.Label, FormControl.Validation, and FormControl.Caption, FormControl.LeadingVisual should also accept the sx prop

Pick a branch
</Text>
<FormControl>
<FormControl.Label>Pick a branch</FormControl.Label>

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Clean ✨

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Love it 💖

@mperrotti
mperrotti merged commit fad43d6 into mainFeb 9, 2022
@mperrotti
mperrotti deleted the mp/fix-form-control-docs branch February 9, 2022 22:00
@primer-cssprimer-css mentioned this pull request Feb 9, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@mperrotti@colebemis