Skip to content

Button size was incorrect. - #1864

Merged
pksjce merged 3 commits into
mainfrom
pk/icon-button-size
Feb 16, 2022
Merged

Button size was incorrect.#1864
pksjce merged 3 commits into
mainfrom
pk/icon-button-size

Conversation

@pksjce

Copy link
Copy Markdown
Contributor

Describe your changes here.

Button size in our storybooks was incorrect for invisible button. Height should be 32px.
Updated padding calculation to reflect that.
Some extra checks to iconLabel and children to fix couple of issues I found while implementing Button2 in memex.

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@pksjce
pksjce requested review from a team and colebemisFebruary 14, 2022 01:13
@changeset-bot

changeset-botBot commented Feb 14, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a13f062

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Feb 14, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js62.6 KB (0%)
dist/browser.umd.js62.99 KB (0%)

@pksjce
pksjce merged commit 8558ae5 into mainFeb 16, 2022
@pksjce
pksjce deleted the pk/icon-button-size branch February 16, 2022 05:56
@primer-cssprimer-css mentioned this pull request Feb 16, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@pksjce@colebemis