Skip to content

Upgrade @primer/behaviors to 1.1.0 - #1877

Merged
siddharthkp merged 2 commits into
mainfrom
siddharth/upgrade-behaviors-1.1.0
Feb 21, 2022
Merged

Upgrade @primer/behaviors to 1.1.0#1877
siddharthkp merged 2 commits into
mainfrom
siddharth/upgrade-behaviors-1.1.0

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Feb 21, 2022

Copy link
Copy Markdown
Member

Release branch: primer/behaviors#59

tl;dr: accessibility improvements for focusTrap

For components that use focus trap, when the component is open, the first item will have focus. The method of opening the menu (mouse click or keypress) does not matter.

Components that use focusTrap:

  • Dialog (and by extension: ConfirmationDialog)
  • AnchoredOverlay (and by extension: ActionMenu, ActionMenu2, DropdownMenu, DropdownMenu2, SelectPanel, Autocomplete)

@siddharthkp
siddharthkp requested review from a team and rezrahFebruary 21, 2022 15:51
@changeset-bot

changeset-botBot commented Feb 21, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 807412c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkpsiddharthkp self-assigned this Feb 21, 2022
@siddharthkpsiddharthkp added dependencies Pull requests that update a dependency file react labels Feb 21, 2022
@github-actions

github-actionsBot commented Feb 21, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js62.62 KB (+0.04% 🔺)
dist/browser.umd.js63.01 KB (+0.04% 🔺)

@siddharthkp
siddharthkp merged commit 50fa988 into mainFeb 21, 2022
@siddharthkp
siddharthkp deleted the siddharth/upgrade-behaviors-1.1.0 branch February 21, 2022 16:09
@primer-cssprimer-css mentioned this pull request Feb 21, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependenciesPull requests that update a dependency filereact

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@rezrah