Skip to content

v35: Update ActionMenu docs to use PropsTable + Fix outdated example - #1931

Merged
siddharthkp merged 1 commit into
next-majorfrom
action-stuff-use-props-table
Mar 7, 2022
Merged

v35: Update ActionMenu docs to use PropsTable + Fix outdated example#1931
siddharthkp merged 1 commit into
next-majorfrom
action-stuff-use-props-table

Conversation

@siddharthkp

Copy link
Copy Markdown
Member

Bonus: Remove the use of ButtonInvisible in example

@siddharthkpsiddharthkp added docs Documentation react skip changeset This change does not need a changelog labels Mar 7, 2022
@siddharthkpsiddharthkp self-assigned this Mar 7, 2022
@siddharthkp
siddharthkp requested review from a team and pksjceMarch 7, 2022 12:21
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0bba1ae

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkp
siddharthkp changed the base branch from main to next-majorMarch 7, 2022 12:23
@siddharthkp
siddharthkp requested review from colebemis and rezrah and removed request for pksjceMarch 7, 2022 12:23
@siddharthkpsiddharthkp changed the title Update ActionMenu docs to use PropsTableUpdate ActionMenu docs to use PropsTable + Fix outdated exampleMar 7, 2022
@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js62.97 KB (-1.98% 🔽)
dist/browser.umd.js63.31 KB (-2.01% 🔽)

@siddharthkpsiddharthkp changed the title Update ActionMenu docs to use PropsTable + Fix outdated examplev35: Update ActionMenu docs to use PropsTable + Fix outdated exampleMar 7, 2022
@siddharthkp
siddharthkp merged commit a88749a into next-majorMar 7, 2022
@siddharthkp
siddharthkp deleted the action-stuff-use-props-table branch March 7, 2022 16:45
colebemis added a commit that referenced this pull request Mar 9, 2022
* prepare integration branch
* Move deprecated components to deprecated folder (#1881)
* prepare integration branch
* Move deprecated components to deprecated folder
* Add subpath exports for deprecated
* Fix up the docs
* Remove dialog and formgroup components
* Fix up all the tests
* Create smooth-cameras-prove.md
Co-authored-by: Reza Rahman <rezrah@github.com>
* Prefix component with types exported from ActionList2 and ActionMenu2 (#1883)
* prefix component name to types
* add changeset
* Update .changeset/actionlist2-actionmenu2-prefix-types.md
Co-authored-by: Cole Bemis <colebemis@github.com>
Co-authored-by: Cole Bemis <colebemis@github.com>
* Export PageLayout from main bundle (#1882)
* Export PageLayout from main bundle
* Create odd-apes-guess.md
* Add PageLayout to sidenav
* Update .changeset/odd-apes-guess.md
Co-authored-by: Rez <rezrah@github.com>
* Update PageLayout docs
Co-authored-by: Rez <rezrah@github.com>
* Update release notes and add deprecation notices for SelectMenu and Dropdown (#1887)
* chore: add deprecation notices for SelectMenu and Dropdown
* fix: deprecated metadata scope in docs
* docs: update release notes for Flex and BorderBox
* docs: update release notes for Position
* chore: add deprecated scope to jsx snippets
* chore: add release notes for Dropdown deprecation
* chore: add Grid release notes
* update release notes
* Deprecate ActionMenu v1 & Promote drafts/ActionMenu v2 (#1895)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* Deprecate ActionMenu - part 1
* Deprecate ActionMenu - part 2
* Promote drafts/ActionMenu2 to main/ActionMenu
* Add changelog
* Add @deprecated on deprecated/ActionMenu
* docs fixed!
* Revert "Deprecate ActionMenu v1 & Promote drafts/ActionMenu v2 (#1895)" (#1896)
This reverts commit f366604.
* Deprecate ActionList v1 & Promote drafts/ActionList v2 (#1893)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* changelog oopsie
* fix docs copy
* Add @deprecated on deprecated/ActionList
* oopsie on the link
* PR feedback on deprecation message
* Deprecate ActionMenu v1 & Promote drafts/ActionMenu v2 (#1897)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* Deprecate ActionMenu - part 1
* Deprecate ActionMenu - part 2
* Promote drafts/ActionMenu2 to main/ActionMenu
* Add changelog
* Add @deprecated on deprecated/ActionMenu
* docs fixed!
* reorder deprecated components alphabetically
* Update deprecation message
* Fix missing icon that only broke on this PR for some reason
* Deprecate DropdownMenu v1 + Remove DropdownMenu v2 (#1898)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* Deprecate ActionMenu - part 1
* Deprecate ActionMenu - part 2
* Promote drafts/ActionMenu2 to main/ActionMenu
* Add changelog
* Add @deprecated on deprecated/ActionMenu
* docs fixed!
* reorder deprecated components alphabetically
* Update deprecation message
* Fix missing icon that only broke on this PR for some reason
* Deprecate DropdownMenu
* Use deprecated Dropdown for theme switcher
* Delete drafts/DropdownMenu2
* Add changelog
* remove debug statement :)
* Fix code examples in changeset (#1901)
* Update .changeset/deprecate-actionlistv1-promote-actionlistv2.md
* Add linebreaks to changesets
* Button deprecation (#1908)
* Move old button to deprecated
* Move Button2 to main bundle
* Add migration docs
* More changes from the checklist
* More deprecation
* Update tests
* Add deprecated details
* Create many-roses-hammer.md
* Update .changeset/many-roses-hammer.md
Co-authored-by: Rez <rezrah@github.com>
* Update many-roses-hammer.md
* Update many-roses-hammer.md
* Update many-roses-hammer.md
Co-authored-by: Rez <rezrah@github.com>
* Deprecate components replaced by FormControl (#1888)
* moves InputField to deprecated package
* adds jsdoc comments for deprecation
* adds changeset
* fixes bad autoformatting in changeset
* fixes import path in test file
* addresses PR feedback
* fixes FormGroup import in test
* addresses PR feedback
* moves tests to depreacted directory
* removes docs header image
* fixes bad formatting in changelog markdown
* Revert "fix merge conflicts with main"
This reverts commit 363bb07, reversing
changes made to f94dcd3.
* Deprecate ChoiceFieldset and ChoiceInputField (#1900)
* deprecates ChoiceFieldset and ChoiceInputField
* update import paths in tests
* addresses more PR feedback
* fixes bad imports in docs
* updates outdated story and test
* moves components to deprecated directory
* fixes broken docs pages
* fixes bad formatting in changelog markdown
* minor tweaks
* moves deprecated notation for ChoiceInputField
* Deprecate legacy Label component (#1889)
* moves Label to deprecated bundle
* graduates Label2 to main bundle
* cleanup
* adds changeset
* fixes tests
* update legacy label snapshots
* Update .changeset/heavy-points-marry.md
Co-authored-by: Rez <rezrah@github.com>
* Update docs/content/Label.mdx
Co-authored-by: Rez <rezrah@github.com>
* addresses pr feedback
* adds Label back to sidebar
* updates tests, removes docs header image
* removes docs header image
* fixes tests
* resets changed test
Co-authored-by: Rez <rezrah@github.com>
* Move remaining deprecated components to src/deprecated (#1917)
* Move deprecated components to src/deprecated
* Fix imports for tests
* Various fixes for v35 (#1916)
* chore: ad-hoc fixes for next-major branch
* fix#1912 docs error
* docs: remove reference to system props
* Create changeset-for-1917.md (#1919)
* Add align prop on ActionMenu.Overlay to pass through to AnchoredOverlay (#1921)
* Allow ActionMenu.Overlay to pass through align prop to AnchoredOverlay
* Add align prop to docs
* Fix AnchoredOverlay docs
* add changeset
* ActionMenu.Button: Fix spacing between text and caret (#1922)
* Fix spacing between Text and caret for ActionMenu.Button
* Improve changelog
* fix margin after design review
* unrelated change
* Update ActionMenu docs (#1931)
* Button: Update active styles + use them for aria-expanded (#1915)
* Add aria-expanded styles for Button
* update default button active styles
* update snapshot for button
* Create healthy-dots-chew.md
* fix broken test following main merging conflict resolution
* Fixes for v35 (#1934)
* fix ActioMenu story
* backfill missing exports in root index
* fix package-lock
* add changeset
* v35: Remove type aliases for ActionList (#1937)
* Replace aliases for ActionList types
* export types for LinkItem
* update selection types
* Next major changeset edits (#1939)
* Remove unnecessary changesets
* Edit changesets
* Fixes for v35 (#1938)
* add missing root exports
* fix invalid export
* add missing fragments to changesets (#1940)
* Update changeset examples (#1941)
* Fix table in changeset
* Add missing comma
* Fix grid changeset
* Update button changeset
Co-authored-by: Reza Rahman <rezrah@github.com>
Co-authored-by: Siddharth Kshetrapal <siddharthkp@github.com>
Co-authored-by: Pavithra Kodmad <pksjce@github.com>
Co-authored-by: Mike Perrotti <mperrotti@github.com>
pksjce added a commit that referenced this pull request Mar 10, 2022
* prepare integration branch
* Move deprecated components to deprecated folder (#1881)
* prepare integration branch
* Move deprecated components to deprecated folder
* Add subpath exports for deprecated
* Fix up the docs
* Remove dialog and formgroup components
* Fix up all the tests
* Create smooth-cameras-prove.md
Co-authored-by: Reza Rahman <rezrah@github.com>
* Prefix component with types exported from ActionList2 and ActionMenu2 (#1883)
* prefix component name to types
* add changeset
* Update .changeset/actionlist2-actionmenu2-prefix-types.md
Co-authored-by: Cole Bemis <colebemis@github.com>
Co-authored-by: Cole Bemis <colebemis@github.com>
* Export PageLayout from main bundle (#1882)
* Export PageLayout from main bundle
* Create odd-apes-guess.md
* Add PageLayout to sidenav
* Update .changeset/odd-apes-guess.md
Co-authored-by: Rez <rezrah@github.com>
* Update PageLayout docs
Co-authored-by: Rez <rezrah@github.com>
* Update release notes and add deprecation notices for SelectMenu and Dropdown (#1887)
* chore: add deprecation notices for SelectMenu and Dropdown
* fix: deprecated metadata scope in docs
* docs: update release notes for Flex and BorderBox
* docs: update release notes for Position
* chore: add deprecated scope to jsx snippets
* chore: add release notes for Dropdown deprecation
* chore: add Grid release notes
* update release notes
* Deprecate ActionMenu v1 & Promote drafts/ActionMenu v2 (#1895)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* Deprecate ActionMenu - part 1
* Deprecate ActionMenu - part 2
* Promote drafts/ActionMenu2 to main/ActionMenu
* Add changelog
* Add @deprecated on deprecated/ActionMenu
* docs fixed!
* Revert "Deprecate ActionMenu v1 & Promote drafts/ActionMenu v2 (#1895)" (#1896)
This reverts commit f366604.
* Deprecate ActionList v1 & Promote drafts/ActionList v2 (#1893)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* changelog oopsie
* fix docs copy
* Add @deprecated on deprecated/ActionList
* oopsie on the link
* PR feedback on deprecation message
* Deprecate ActionMenu v1 & Promote drafts/ActionMenu v2 (#1897)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* Deprecate ActionMenu - part 1
* Deprecate ActionMenu - part 2
* Promote drafts/ActionMenu2 to main/ActionMenu
* Add changelog
* Add @deprecated on deprecated/ActionMenu
* docs fixed!
* reorder deprecated components alphabetically
* Update deprecation message
* Fix missing icon that only broke on this PR for some reason
* Deprecate DropdownMenu v1 + Remove DropdownMenu v2 (#1898)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* Deprecate ActionMenu - part 1
* Deprecate ActionMenu - part 2
* Promote drafts/ActionMenu2 to main/ActionMenu
* Add changelog
* Add @deprecated on deprecated/ActionMenu
* docs fixed!
* reorder deprecated components alphabetically
* Update deprecation message
* Fix missing icon that only broke on this PR for some reason
* Deprecate DropdownMenu
* Use deprecated Dropdown for theme switcher
* Delete drafts/DropdownMenu2
* Add changelog
* remove debug statement :)
* Fix code examples in changeset (#1901)
* Update .changeset/deprecate-actionlistv1-promote-actionlistv2.md
* Add linebreaks to changesets
* Button deprecation (#1908)
* Move old button to deprecated
* Move Button2 to main bundle
* Add migration docs
* More changes from the checklist
* More deprecation
* Update tests
* Add deprecated details
* Create many-roses-hammer.md
* Update .changeset/many-roses-hammer.md
Co-authored-by: Rez <rezrah@github.com>
* Update many-roses-hammer.md
* Update many-roses-hammer.md
* Update many-roses-hammer.md
Co-authored-by: Rez <rezrah@github.com>
* Deprecate components replaced by FormControl (#1888)
* moves InputField to deprecated package
* adds jsdoc comments for deprecation
* adds changeset
* fixes bad autoformatting in changeset
* fixes import path in test file
* addresses PR feedback
* fixes FormGroup import in test
* addresses PR feedback
* moves tests to depreacted directory
* removes docs header image
* fixes bad formatting in changelog markdown
* Revert "fix merge conflicts with main"
This reverts commit 363bb07, reversing
changes made to f94dcd3.
* Deprecate ChoiceFieldset and ChoiceInputField (#1900)
* deprecates ChoiceFieldset and ChoiceInputField
* update import paths in tests
* addresses more PR feedback
* fixes bad imports in docs
* updates outdated story and test
* moves components to deprecated directory
* fixes broken docs pages
* fixes bad formatting in changelog markdown
* minor tweaks
* moves deprecated notation for ChoiceInputField
* Deprecate legacy Label component (#1889)
* moves Label to deprecated bundle
* graduates Label2 to main bundle
* cleanup
* adds changeset
* fixes tests
* update legacy label snapshots
* Update .changeset/heavy-points-marry.md
Co-authored-by: Rez <rezrah@github.com>
* Update docs/content/Label.mdx
Co-authored-by: Rez <rezrah@github.com>
* addresses pr feedback
* adds Label back to sidebar
* updates tests, removes docs header image
* removes docs header image
* fixes tests
* resets changed test
Co-authored-by: Rez <rezrah@github.com>
* Move remaining deprecated components to src/deprecated (#1917)
* Move deprecated components to src/deprecated
* Fix imports for tests
* Various fixes for v35 (#1916)
* chore: ad-hoc fixes for next-major branch
* fix#1912 docs error
* docs: remove reference to system props
* Create changeset-for-1917.md (#1919)
* Add align prop on ActionMenu.Overlay to pass through to AnchoredOverlay (#1921)
* Allow ActionMenu.Overlay to pass through align prop to AnchoredOverlay
* Add align prop to docs
* Fix AnchoredOverlay docs
* add changeset
* ActionMenu.Button: Fix spacing between text and caret (#1922)
* Fix spacing between Text and caret for ActionMenu.Button
* Improve changelog
* fix margin after design review
* unrelated change
* Update ActionMenu docs (#1931)
* Button: Update active styles + use them for aria-expanded (#1915)
* Add aria-expanded styles for Button
* update default button active styles
* update snapshot for button
* Create healthy-dots-chew.md
* fix broken test following main merging conflict resolution
* Fixes for v35 (#1934)
* fix ActioMenu story
* backfill missing exports in root index
* fix package-lock
* add changeset
* v35: Remove type aliases for ActionList (#1937)
* Replace aliases for ActionList types
* export types for LinkItem
* update selection types
* Next major changeset edits (#1939)
* Remove unnecessary changesets
* Edit changesets
* Fixes for v35 (#1938)
* add missing root exports
* fix invalid export
* add missing fragments to changesets (#1940)
* Update changeset examples (#1941)
* Fix table in changeset
* Add missing comma
* Fix grid changeset
* Update button changeset
Co-authored-by: Reza Rahman <rezrah@github.com>
Co-authored-by: Siddharth Kshetrapal <siddharthkp@github.com>
Co-authored-by: Pavithra Kodmad <pksjce@github.com>
Co-authored-by: Mike Perrotti <mperrotti@github.com>
mperrotti added a commit that referenced this pull request Mar 16, 2022
* prepare integration branch
* Move deprecated components to deprecated folder (#1881)
* prepare integration branch
* Move deprecated components to deprecated folder
* Add subpath exports for deprecated
* Fix up the docs
* Remove dialog and formgroup components
* Fix up all the tests
* Create smooth-cameras-prove.md
Co-authored-by: Reza Rahman <rezrah@github.com>
* Prefix component with types exported from ActionList2 and ActionMenu2 (#1883)
* prefix component name to types
* add changeset
* Update .changeset/actionlist2-actionmenu2-prefix-types.md
Co-authored-by: Cole Bemis <colebemis@github.com>
Co-authored-by: Cole Bemis <colebemis@github.com>
* Export PageLayout from main bundle (#1882)
* Export PageLayout from main bundle
* Create odd-apes-guess.md
* Add PageLayout to sidenav
* Update .changeset/odd-apes-guess.md
Co-authored-by: Rez <rezrah@github.com>
* Update PageLayout docs
Co-authored-by: Rez <rezrah@github.com>
* Update release notes and add deprecation notices for SelectMenu and Dropdown (#1887)
* chore: add deprecation notices for SelectMenu and Dropdown
* fix: deprecated metadata scope in docs
* docs: update release notes for Flex and BorderBox
* docs: update release notes for Position
* chore: add deprecated scope to jsx snippets
* chore: add release notes for Dropdown deprecation
* chore: add Grid release notes
* update release notes
* Deprecate ActionMenu v1 & Promote drafts/ActionMenu v2 (#1895)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* Deprecate ActionMenu - part 1
* Deprecate ActionMenu - part 2
* Promote drafts/ActionMenu2 to main/ActionMenu
* Add changelog
* Add @deprecated on deprecated/ActionMenu
* docs fixed!
* Revert "Deprecate ActionMenu v1 & Promote drafts/ActionMenu v2 (#1895)" (#1896)
This reverts commit f366604.
* Deprecate ActionList v1 & Promote drafts/ActionList v2 (#1893)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* changelog oopsie
* fix docs copy
* Add @deprecated on deprecated/ActionList
* oopsie on the link
* PR feedback on deprecation message
* Deprecate ActionMenu v1 & Promote drafts/ActionMenu v2 (#1897)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* Deprecate ActionMenu - part 1
* Deprecate ActionMenu - part 2
* Promote drafts/ActionMenu2 to main/ActionMenu
* Add changelog
* Add @deprecated on deprecated/ActionMenu
* docs fixed!
* reorder deprecated components alphabetically
* Update deprecation message
* Fix missing icon that only broke on this PR for some reason
* Deprecate DropdownMenu v1 + Remove DropdownMenu v2 (#1898)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* Deprecate ActionMenu - part 1
* Deprecate ActionMenu - part 2
* Promote drafts/ActionMenu2 to main/ActionMenu
* Add changelog
* Add @deprecated on deprecated/ActionMenu
* docs fixed!
* reorder deprecated components alphabetically
* Update deprecation message
* Fix missing icon that only broke on this PR for some reason
* Deprecate DropdownMenu
* Use deprecated Dropdown for theme switcher
* Delete drafts/DropdownMenu2
* Add changelog
* remove debug statement :)
* Fix code examples in changeset (#1901)
* Update .changeset/deprecate-actionlistv1-promote-actionlistv2.md
* Add linebreaks to changesets
* Button deprecation (#1908)
* Move old button to deprecated
* Move Button2 to main bundle
* Add migration docs
* More changes from the checklist
* More deprecation
* Update tests
* Add deprecated details
* Create many-roses-hammer.md
* Update .changeset/many-roses-hammer.md
Co-authored-by: Rez <rezrah@github.com>
* Update many-roses-hammer.md
* Update many-roses-hammer.md
* Update many-roses-hammer.md
Co-authored-by: Rez <rezrah@github.com>
* Deprecate components replaced by FormControl (#1888)
* moves InputField to deprecated package
* adds jsdoc comments for deprecation
* adds changeset
* fixes bad autoformatting in changeset
* fixes import path in test file
* addresses PR feedback
* fixes FormGroup import in test
* addresses PR feedback
* moves tests to depreacted directory
* removes docs header image
* fixes bad formatting in changelog markdown
* Revert "fix merge conflicts with main"
This reverts commit 363bb07, reversing
changes made to f94dcd3.
* Deprecate ChoiceFieldset and ChoiceInputField (#1900)
* deprecates ChoiceFieldset and ChoiceInputField
* update import paths in tests
* addresses more PR feedback
* fixes bad imports in docs
* updates outdated story and test
* moves components to deprecated directory
* fixes broken docs pages
* fixes bad formatting in changelog markdown
* minor tweaks
* moves deprecated notation for ChoiceInputField
* Deprecate legacy Label component (#1889)
* moves Label to deprecated bundle
* graduates Label2 to main bundle
* cleanup
* adds changeset
* fixes tests
* update legacy label snapshots
* Update .changeset/heavy-points-marry.md
Co-authored-by: Rez <rezrah@github.com>
* Update docs/content/Label.mdx
Co-authored-by: Rez <rezrah@github.com>
* addresses pr feedback
* adds Label back to sidebar
* updates tests, removes docs header image
* removes docs header image
* fixes tests
* resets changed test
Co-authored-by: Rez <rezrah@github.com>
* Move remaining deprecated components to src/deprecated (#1917)
* Move deprecated components to src/deprecated
* Fix imports for tests
* Various fixes for v35 (#1916)
* chore: ad-hoc fixes for next-major branch
* fix#1912 docs error
* docs: remove reference to system props
* Create changeset-for-1917.md (#1919)
* Add align prop on ActionMenu.Overlay to pass through to AnchoredOverlay (#1921)
* Allow ActionMenu.Overlay to pass through align prop to AnchoredOverlay
* Add align prop to docs
* Fix AnchoredOverlay docs
* add changeset
* ActionMenu.Button: Fix spacing between text and caret (#1922)
* Fix spacing between Text and caret for ActionMenu.Button
* Improve changelog
* fix margin after design review
* unrelated change
* Update ActionMenu docs (#1931)
* Button: Update active styles + use them for aria-expanded (#1915)
* Add aria-expanded styles for Button
* update default button active styles
* update snapshot for button
* Create healthy-dots-chew.md
* fix broken test following main merging conflict resolution
* Fixes for v35 (#1934)
* fix ActioMenu story
* backfill missing exports in root index
* fix package-lock
* add changeset
* v35: Remove type aliases for ActionList (#1937)
* Replace aliases for ActionList types
* export types for LinkItem
* update selection types
* Next major changeset edits (#1939)
* Remove unnecessary changesets
* Edit changesets
* Fixes for v35 (#1938)
* add missing root exports
* fix invalid export
* add missing fragments to changesets (#1940)
* Update changeset examples (#1941)
* Fix table in changeset
* Add missing comma
* Fix grid changeset
* Update button changeset
* rms deprecated components and layout tweaks
* updates snaps
* adds changeset
* Update title and componentId of deprecated components
* Add missing component ids
* Remove link button docs
* Update dialog props
* Update nav.yml
* Update live code scope
Co-authored-by: Reza Rahman <rezrah@github.com>
Co-authored-by: Siddharth Kshetrapal <siddharthkp@github.com>
Co-authored-by: Cole Bemis <colebemis@github.com>
Co-authored-by: Pavithra Kodmad <pksjce@github.com>
pksjce added a commit that referenced this pull request Mar 17, 2022
* prepare integration branch
* Move deprecated components to deprecated folder (#1881)
* prepare integration branch
* Move deprecated components to deprecated folder
* Add subpath exports for deprecated
* Fix up the docs
* Remove dialog and formgroup components
* Fix up all the tests
* Create smooth-cameras-prove.md
Co-authored-by: Reza Rahman <rezrah@github.com>
* Prefix component with types exported from ActionList2 and ActionMenu2 (#1883)
* prefix component name to types
* add changeset
* Update .changeset/actionlist2-actionmenu2-prefix-types.md
Co-authored-by: Cole Bemis <colebemis@github.com>
Co-authored-by: Cole Bemis <colebemis@github.com>
* Export PageLayout from main bundle (#1882)
* Export PageLayout from main bundle
* Create odd-apes-guess.md
* Add PageLayout to sidenav
* Update .changeset/odd-apes-guess.md
Co-authored-by: Rez <rezrah@github.com>
* Update PageLayout docs
Co-authored-by: Rez <rezrah@github.com>
* Update release notes and add deprecation notices for SelectMenu and Dropdown (#1887)
* chore: add deprecation notices for SelectMenu and Dropdown
* fix: deprecated metadata scope in docs
* docs: update release notes for Flex and BorderBox
* docs: update release notes for Position
* chore: add deprecated scope to jsx snippets
* chore: add release notes for Dropdown deprecation
* chore: add Grid release notes
* update release notes
* Deprecate ActionMenu v1 & Promote drafts/ActionMenu v2 (#1895)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* Deprecate ActionMenu - part 1
* Deprecate ActionMenu - part 2
* Promote drafts/ActionMenu2 to main/ActionMenu
* Add changelog
* Add @deprecated on deprecated/ActionMenu
* docs fixed!
* Revert "Deprecate ActionMenu v1 & Promote drafts/ActionMenu v2 (#1895)" (#1896)
This reverts commit f366604.
* Deprecate ActionList v1 & Promote drafts/ActionList v2 (#1893)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* changelog oopsie
* fix docs copy
* Add @deprecated on deprecated/ActionList
* oopsie on the link
* PR feedback on deprecation message
* Deprecate ActionMenu v1 & Promote drafts/ActionMenu v2 (#1897)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* Deprecate ActionMenu - part 1
* Deprecate ActionMenu - part 2
* Promote drafts/ActionMenu2 to main/ActionMenu
* Add changelog
* Add @deprecated on deprecated/ActionMenu
* docs fixed!
* reorder deprecated components alphabetically
* Update deprecation message
* Fix missing icon that only broke on this PR for some reason
* Deprecate DropdownMenu v1 + Remove DropdownMenu v2 (#1898)
* Deprecate ActionList v1
* Promote drafts/ActionList2 to main/ActionList
* Add changelog
* Undo package-lock change
* update ActionList import for Menu2 docs
* Deprecate ActionMenu - part 1
* Deprecate ActionMenu - part 2
* Promote drafts/ActionMenu2 to main/ActionMenu
* Add changelog
* Add @deprecated on deprecated/ActionMenu
* docs fixed!
* reorder deprecated components alphabetically
* Update deprecation message
* Fix missing icon that only broke on this PR for some reason
* Deprecate DropdownMenu
* Use deprecated Dropdown for theme switcher
* Delete drafts/DropdownMenu2
* Add changelog
* remove debug statement :)
* Fix code examples in changeset (#1901)
* Update .changeset/deprecate-actionlistv1-promote-actionlistv2.md
* Add linebreaks to changesets
* Button deprecation (#1908)
* Move old button to deprecated
* Move Button2 to main bundle
* Add migration docs
* More changes from the checklist
* More deprecation
* Update tests
* Add deprecated details
* Create many-roses-hammer.md
* Update .changeset/many-roses-hammer.md
Co-authored-by: Rez <rezrah@github.com>
* Update many-roses-hammer.md
* Update many-roses-hammer.md
* Update many-roses-hammer.md
Co-authored-by: Rez <rezrah@github.com>
* Deprecate components replaced by FormControl (#1888)
* moves InputField to deprecated package
* adds jsdoc comments for deprecation
* adds changeset
* fixes bad autoformatting in changeset
* fixes import path in test file
* addresses PR feedback
* fixes FormGroup import in test
* addresses PR feedback
* moves tests to depreacted directory
* removes docs header image
* fixes bad formatting in changelog markdown
* Revert "fix merge conflicts with main"
This reverts commit 363bb07, reversing
changes made to f94dcd3.
* Deprecate ChoiceFieldset and ChoiceInputField (#1900)
* deprecates ChoiceFieldset and ChoiceInputField
* update import paths in tests
* addresses more PR feedback
* fixes bad imports in docs
* updates outdated story and test
* moves components to deprecated directory
* fixes broken docs pages
* fixes bad formatting in changelog markdown
* minor tweaks
* moves deprecated notation for ChoiceInputField
* Deprecate legacy Label component (#1889)
* moves Label to deprecated bundle
* graduates Label2 to main bundle
* cleanup
* adds changeset
* fixes tests
* update legacy label snapshots
* Update .changeset/heavy-points-marry.md
Co-authored-by: Rez <rezrah@github.com>
* Update docs/content/Label.mdx
Co-authored-by: Rez <rezrah@github.com>
* addresses pr feedback
* adds Label back to sidebar
* updates tests, removes docs header image
* removes docs header image
* fixes tests
* resets changed test
Co-authored-by: Rez <rezrah@github.com>
* Move remaining deprecated components to src/deprecated (#1917)
* Move deprecated components to src/deprecated
* Fix imports for tests
* Various fixes for v35 (#1916)
* chore: ad-hoc fixes for next-major branch
* fix#1912 docs error
* docs: remove reference to system props
* Create changeset-for-1917.md (#1919)
* Add align prop on ActionMenu.Overlay to pass through to AnchoredOverlay (#1921)
* Allow ActionMenu.Overlay to pass through align prop to AnchoredOverlay
* Add align prop to docs
* Fix AnchoredOverlay docs
* add changeset
* ActionMenu.Button: Fix spacing between text and caret (#1922)
* Fix spacing between Text and caret for ActionMenu.Button
* Improve changelog
* fix margin after design review
* unrelated change
* Update ActionMenu docs (#1931)
* Button: Update active styles + use them for aria-expanded (#1915)
* Add aria-expanded styles for Button
* update default button active styles
* update snapshot for button
* Create healthy-dots-chew.md
* fix broken test following main merging conflict resolution
* Fixes for v35 (#1934)
* fix ActioMenu story
* backfill missing exports in root index
* fix package-lock
* add changeset
* v35: Remove type aliases for ActionList (#1937)
* Replace aliases for ActionList types
* export types for LinkItem
* update selection types
* Next major changeset edits (#1939)
* Remove unnecessary changesets
* Edit changesets
* Fixes for v35 (#1938)
* add missing root exports
* fix invalid export
* add missing fragments to changesets (#1940)
* Update changeset examples (#1941)
* Fix table in changeset
* Add missing comma
* Fix grid changeset
* Update button changeset
* rms deprecated components and layout tweaks
* updates snaps
* adds changeset
* Update title and componentId of deprecated components
* Add missing component ids
* Remove link button docs
* Update dialog props
* Update nav.yml
* Update live code scope
Co-authored-by: Reza Rahman <rezrah@github.com>
Co-authored-by: Siddharth Kshetrapal <siddharthkp@github.com>
Co-authored-by: Cole Bemis <colebemis@github.com>
Co-authored-by: Pavithra Kodmad <pksjce@github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

docsDocumentationreactskip changesetThis change does not need a changelog

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@colebemis