Skip to content

Forward ref to FormControl - #1949

Merged
colebemis merged 3 commits into
mainfrom
form-control-ref
Mar 10, 2022
Merged

Forward ref to FormControl#1949
colebemis merged 3 commits into
mainfrom
form-control-ref

Conversation

@colebemis

@colebemiscolebemis commented Mar 10, 2022

Copy link
Copy Markdown
Contributor

We recently deprecated FormGroup in favor of FormControl. However, @dusave ran into issues migrating from FormGroup to FormControl because FormControl didn't accept a ref prop. Now it does :)

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@changeset-bot

changeset-botBot commented Mar 10, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 48d26fe

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Mar 10, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js63.18 KB (-0.01% 🔽)
dist/browser.umd.js63.53 KB (+0.02% 🔺)

@colebemis
colebemis marked this pull request as ready for review March 10, 2022 22:01
@colebemis
colebemis requested review from a team and mperrottiMarch 10, 2022 22:01

@dusavedusave left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

👏🏻 Thanks for the quick patch!

@colebemis
colebemis merged commit e430bd8 into mainMar 10, 2022
@colebemis
colebemis deleted the form-control-ref branch March 10, 2022 23:23
@primer-cssprimer-css mentioned this pull request Mar 10, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@colebemis@pksjce@dusave