Skip to content

Bump primer/primitives to 7.5.1 - #1961

Merged
simurai merged 9 commits into
mainfrom
bump-primitives-7.5.0
Mar 16, 2022
Merged

Bump primer/primitives to 7.5.1#1961
simurai merged 9 commits into
mainfrom
bump-primitives-7.5.0

Conversation

@simurai

@simuraisimurai commented Mar 15, 2022

Copy link
Copy Markdown
Contributor

This bumps primer/primitives to 7.5.1.

It's mostly needed for the new open and closed color roles added in 7.5.0

What should reviewers focus on

  • Are there any more changes needed? There shouldn't be any breaking changes in v7.1.1...v7.5.1

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

@changeset-bot

changeset-botBot commented Mar 15, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 92904db

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Mar 15, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js63.39 KB (+0.25% 🔺)
dist/browser.umd.js63.74 KB (+0.25% 🔺)

@simurai
simurai marked this pull request as ready for review March 15, 2022 06:45
@simurai
simurai requested review from a team and pksjceMarch 15, 2022 06:45
@simurai

simurai commented Mar 15, 2022

Copy link
Copy Markdown
ContributorAuthor

The following test was failing:

 FAIL src/__tests__/themePreval.test.ts (16.005 s)
● snapshot theme-preval.js
ENOENT: no such file or directory, open '/workspaces/react/lib/theme-preval.js'
4 | test('snapshot theme-preval.js', () => {
5 | // Snapshot test the build output of theme-preval.js to catch any unexpected changes
> 6 | const themePrevalFile = fs.readFileSync(path.resolve(__dirname, '../../lib/theme-preval.js'), 'utf8')
| ^
7 | expect(themePrevalFile).toMatchSnapshot()
8 | })
9 |
at Object.<anonymous> (src/__tests__/themePreval.test.ts:6:30)

Updating the date here and then running test:update

"test:update": "npm run test -- --updateSnapshot",
seemed to fix it. See 3f9b63e.

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks, @simurai!

@simurai
simurai enabled auto-merge (squash) March 16, 2022 04:24
@simurai
simurai merged commit 767d416 into mainMar 16, 2022
@simurai
simurai deleted the bump-primitives-7.5.0 branch March 16, 2022 04:30
@primer-cssprimer-css mentioned this pull request Mar 16, 2022
@simuraisimurai mentioned this pull request Mar 22, 2022
6 tasks
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@simurai@colebemis