Skip to content

Add overlay props to Autocomplete.Overlay - #1967

Merged
pksjce merged 3 commits into
mainfrom
pk/autocomplete-overlay-bug
Mar 17, 2022
Merged

Add overlay props to Autocomplete.Overlay#1967
pksjce merged 3 commits into
mainfrom
pk/autocomplete-overlay-bug

Conversation

@pksjce

@pksjcepksjce commented Mar 16, 2022

Copy link
Copy Markdown
Contributor

Describe your changes here.

Add overlay props to Autocomplete.Overlay

Closes#1951

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

@pksjce
pksjce requested review from a team and rezrahMarch 16, 2022 09:34
@changeset-bot

changeset-botBot commented Mar 16, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 63272bb

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pksjce
pksjceforce-pushed the pk/autocomplete-overlay-bug branch from c408554 to 042d372CompareMarch 16, 2022 09:35
@github-actions

github-actionsBot commented Mar 16, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js63.2 KB (+0.02% 🔺)
dist/browser.umd.js63.56 KB (+0.02% 🔺)

if (autocompleteContext === null) {
throw new Error('AutocompleteContext returned null values')
}
const overlayProps = {...oldOverlayProps, ...newOverlayProps}

@siddharthkpsiddharthkpMar 16, 2022

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

backwards compatibility! love to see it!

@pksjce
pksjceforce-pushed the pk/autocomplete-overlay-bug branch from 042d372 to 8584557CompareMarch 17, 2022 04:10
@pksjcepksjce changed the title WIP: Add overlay props to Autocomplete.OverlayAdd overlay props to Autocomplete.OverlayMar 17, 2022
@pksjce
pksjce enabled auto-merge (squash) March 17, 2022 04:11
@pksjce
pksjce merged commit c83a06f into mainMar 17, 2022
@pksjce
pksjce deleted the pk/autocomplete-overlay-bug branch March 17, 2022 04:16
@primer-cssprimer-css mentioned this pull request Mar 17, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Update Autocomplete.Overlay props API

2 participants

@pksjce@siddharthkp