Skip to content

ActionMenu: Fix styles for windows high contrast mode - #1970

Merged
siddharthkp merged 7 commits into
mainfrom
siddharthkp/actionmenu-high-contrast
Mar 17, 2022
Merged

ActionMenu: Fix styles for windows high contrast mode#1970
siddharthkp merged 7 commits into
mainfrom
siddharthkp/actionmenu-high-contrast

Conversation

@siddharthkp

Copy link
Copy Markdown
Member
  • Add outline for Overlay
  • Add focus styles for Button
  • Update code comments to one consistent message

@siddharthkp
siddharthkp requested review from a team and mperrottiMarch 16, 2022 16:54
@siddharthkpsiddharthkp changed the title ActionMenu: Support for windows high contrastActionMenu: Support for windows high contrast modeMar 16, 2022
@siddharthkpsiddharthkp changed the title ActionMenu: Support for windows high contrast modeActionMenu: Fix styles for windows high contrast modeMar 16, 2022
@siddharthkp
siddharthkp requested review from colebemis and pksjce and removed request for mperrottiMarch 16, 2022 16:54
@siddharthkpsiddharthkp self-assigned this Mar 16, 2022
@changeset-bot

changeset-botBot commented Mar 16, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 20f08ac

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Mar 17, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js63.24 KB (+0.07% 🔺)
dist/browser.umd.js63.59 KB (+0.07% 🔺)

@siddharthkp
siddharthkp merged commit 3b23604 into mainMar 17, 2022
@siddharthkp
siddharthkp deleted the siddharthkp/actionmenu-high-contrast branch March 17, 2022 15:55
@primer-cssprimer-css mentioned this pull request Mar 17, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@colebemis