Uh oh!
There was an error while loading. Please reload this page.
fix: checkbox to useIsomorphicLayoutEffect - #2116
Conversation
🦋 Changeset detectedLatest commit: 1d57e8e The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
mperrotti
commented
Jun 23, 2022
@Swiftwork - Could you open a PR as a branch directly off of this repo? Org tokens aren't propagating or being respected, so deployments aren't working for PRs from a forked repo. |
mperrotti
commented
Jun 28, 2022
@Swiftwork - Since you're an external contributory, I cherry-picked your commits and opened a PR directly off this repo #2154 Thank you! |
Swiftwork
commented
Jun 28, 2022
@Swiftwork thank you for doing the cherry-picking. I was a little confused by the request to open a branch on this PR without contributor access. It would be good to in the future allow for PR as there are a few more issues that have been detected unless you prefer me to open issue tickets instead? |
Describe your changes here.
Closes#2111
Merge checklist