Skip to content

SelectPanel: Accessibility remediation - #2138

Merged
hectahertz merged 45 commits into
mainfrom
primer/issues/1021/reeact-select-panel
Aug 25, 2022
Merged

SelectPanel: Accessibility remediation#2138
hectahertz merged 45 commits into
mainfrom
primer/issues/1021/reeact-select-panel

Conversation

@bolonio

@boloniobolonio commented Jun 23, 2022

Copy link
Copy Markdown

Co-authored-by: @hectahertz

Context

This PR addresses the changes derived from:

The Accessibility engineering review has more details on the reasoning behind all changes.

Screenshots

BeforeAfter
Screenshot of the previous version of the componentScreenshot of the proposed layout for the component, which adds a title, close button and save/cancel buttons

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

@changeset-bot

changeset-botBot commented Jun 23, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b828d90

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Jun 23, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js76.26 KB (+0.85% 🔺)
dist/browser.umd.js76.87 KB (+0.82% 🔺)

@bolonio
bolonio temporarily deployed to github-pages June 23, 2022 12:10 Inactive
@bolonio
bolonio temporarily deployed to github-pages June 23, 2022 13:59 Inactive
@bolonio
bolonioforce-pushed the primer/issues/1021/reeact-select-panel branch from d0d1eca to f851e04CompareJune 23, 2022 14:02
@bolonio
bolonio temporarily deployed to github-pages June 23, 2022 14:07 Inactive
@bolonio
bolonio temporarily deployed to github-pages June 23, 2022 14:31 Inactive
@boloniobolonio changed the title Primer React SelectPanel - a11y engineering review implementationSelectPanel - a11y engineering review implementationJun 23, 2022
@hectahertz
hectahertz temporarily deployed to github-pages August 19, 2022 16:02 Inactive
@hectahertz
hectahertz temporarily deployed to github-pages August 22, 2022 13:47 Inactive

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Awesome work! Thank you, @hectahertz. Looks like there are a few linting errors that need to be fixed. Approving in advance 👍

@colebemiscolebemis changed the title SelectPanel: Accessibility remediation (breaking changes)SelectPanel: Accessibility remediationAug 22, 2022
@hectahertz
hectahertz temporarily deployed to github-pages August 23, 2022 09:04 Inactive
@hectahertz
hectahertz temporarily deployed to github-pages August 23, 2022 10:01 Inactive
@hectahertz
hectahertz temporarily deployed to github-pages August 24, 2022 15:15 Inactive
@hectahertz
hectahertz temporarily deployed to github-pages August 25, 2022 17:27 Inactive
@hectahertz
hectahertz temporarily deployed to github-pages August 25, 2022 18:27 Inactive
@hectahertz
hectahertz temporarily deployed to github-pages August 25, 2022 19:50 Inactive
@hectahertz
hectahertz merged commit ace38af into mainAug 25, 2022
@hectahertz
hectahertz deleted the primer/issues/1021/reeact-select-panel branch August 25, 2022 19:54
@primer-cssprimer-css mentioned this pull request Aug 25, 2022
broccolinisoup added a commit that referenced this pull request Sep 14, 2022
broccolinisoup added a commit that referenced this pull request Sep 20, 2022
* Revert "SelectPanel: Accessibility remediation (#2138)"
This reverts commit ace38af.
* add changeset
* Update .changeset/stupid-knives-arrive.md
Co-authored-by: Brendan Forster <github@brendanforster.com>
Co-authored-by: Brendan Forster <github@brendanforster.com>
@siddharthkp
siddharthkp restored the primer/issues/1021/reeact-select-panel branch January 20, 2023 11:17
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants

@bolonio@vdepizzol@hectahertz@camchenry@siddharthkp@colebemis