Uh oh!
There was an error while loading. Please reload this page.
Support ReactNode in ActionMenu.Overlay - #2168
Merged
siddharthkp merged 4 commits intoJul 26, 2022
Merged
Conversation
🦋 Changeset detectedLatest commit: a16d79c The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
siddharthkp
commented
Jul 20, 2022
Member
This looks pretty safe, can we test this with a canary build in memex just to be sure? |
jdrush89
commented
Jul 26, 2022
ContributorAuthor
@siddharthkp sure, what is the process for that? |
Member
Ugh, canary release workflow was skipped on fork. Let's merge this into a branch in primer/react and continue from there. Update: Creating a new pull request here: #2188 and invited you to the primer/react repository |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for freeto join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.

Describe your changes here.
Add support for ReactNode as the child type in ActionMenu.Overlay so that conditional children can be rendered within an overlay.
Closes#2067
Screenshots
No visual changes
Merge checklist
Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.