Skip to content

Overlay documentation fixes - #2221

Merged
josepmartins merged 3 commits into
mainfrom
josepmartins/docs-overlay-fixes
Aug 8, 2022
Merged

Overlay documentation fixes#2221
josepmartins merged 3 commits into
mainfrom
josepmartins/docs-overlay-fixes

Conversation

@josepmartins

Copy link
Copy Markdown
Contributor

Describe your changes here.

Tiny fixes on the Overlay documentation to avoid confusion, see #1971 (comment) or https://github.slack.com/archives/C01L618AEP9/p1658398312402499@maximedegreve

  • Add an AnchoredOverlay link in the Overlay#Positioning documentation section.
  • Add an AnchoredOverlay item in the navigation.
  • Add/remove missing/non-existing props in the Overlay documentation.

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@josepmartins
josepmartins requested review from a team and mperrottiAugust 8, 2022 09:08
@changeset-bot

changeset-botBot commented Aug 8, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 6b0e19a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@josepmartinsjosepmartins changed the title Josepmartins/docs overlay fixesOverlay documentation fixesAug 8, 2022
@github-actions

github-actionsBot commented Aug 8, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js72.93 KB (0%)
dist/browser.umd.js73.3 KB (0%)

@josepmartins
josepmartins temporarily deployed to github-pages August 8, 2022 09:20 Inactive
@josepmartins
josepmartins temporarily deployed to github-pages August 8, 2022 09:21 Inactive

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I approve this message, ship it :shipit:

@maximedegreve

Copy link
Copy Markdown
Contributor

🎉

@josepmartins
josepmartins merged commit 9ce6493 into mainAug 8, 2022
@josepmartins
josepmartins deleted the josepmartins/docs-overlay-fixes branch August 8, 2022 10:06
@primer-cssprimer-css mentioned this pull request Aug 8, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@josepmartins@maximedegreve@siddharthkp