Skip to content

feat(PageLayout): add support for labeling through aria-label, aria-labelledby - #2294

Merged
joshblack merged 7 commits into
mainfrom
landmark-page-layout
Sep 2, 2022
Merged

feat(PageLayout): add support for labeling through aria-label, aria-labelledby#2294
joshblack merged 7 commits into
mainfrom
landmark-page-layout

Conversation

@joshblack

@joshblackjoshblack commented Aug 31, 2022

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/1254

This PR adds in support for labelling PageLayout.Header, PageLayout.Content, and PageLayout.Footer through either aria-label or aria-labelledby.

It also provides additional information in the PageLayout docs for these landmark roles through a new Accessibility section.

@changeset-bot

changeset-botBot commented Aug 31, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 8a10923

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Aug 31, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js76.29 KB (+0.05% 🔺)
dist/browser.umd.js76.9 KB (+0.05% 🔺)

@joshblack
joshblack temporarily deployed to github-pages August 31, 2022 15:33 Inactive
@joshblackjoshblack changed the title Landmark page layoutfeat(PageLayout): add support for labeling through aria-label, aria-labelledbyAug 31, 2022
@joshblack
joshblack marked this pull request as ready for review August 31, 2022 15:52
@joshblack
joshblack requested review from a team, colebemis and rezrahAugust 31, 2022 15:52
@joshblack
joshblack temporarily deployed to github-pages August 31, 2022 15:55 Inactive
Comment threaddocs/content/PageLayout.mdx Outdated

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great!

Co-authored-by: Cole Bemis <colebemis@github.com>
@joshblack
joshblack temporarily deployed to github-pages August 31, 2022 16:27 Inactive

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks awesome! 🤩

@hectahertzhectahertz left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Great work @joshblack! 🎉

@joshblack
joshblack enabled auto-merge (squash) September 2, 2022 14:36
@joshblack
joshblack temporarily deployed to github-pages September 2, 2022 14:43 Inactive
@joshblack
joshblack merged commit 4536b87 into mainSep 2, 2022
@joshblack
joshblack deleted the landmark-page-layout branch September 2, 2022 14:43
@primer-cssprimer-css mentioned this pull request Sep 2, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@joshblack@broccolinisoup@colebemis@hectahertz