Skip to content

Adds size prop to the SegmentedControl - #2347

Merged
mperrotti merged 4 commits into
mainfrom
mp/add-segmentedcontrol-size-variant
Sep 19, 2022
Merged

Adds size prop to the SegmentedControl#2347
mperrotti merged 4 commits into
mainfrom
mp/add-segmentedcontrol-size-variant

Conversation

@mperrotti

Copy link
Copy Markdown
Contributor

Adds a "small" size option to the SegmentedControl.

Closes#2267

Screenshots

Screen Shot 2022-09-19 at 2 27 27 PM

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@mperrotti
mperrotti requested review from a team and siddharthkpSeptember 19, 2022 18:27
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1b29455

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js76.39 KB (+0.07% 🔺)
dist/browser.umd.js77.05 KB (+0.07% 🔺)

@mperrotti
mperrotti temporarily deployed to github-pages September 19, 2022 18:34 Inactive

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🚢 !

@mperrotti
mperrotti merged commit 09a5d80 into mainSep 19, 2022
@mperrotti
mperrotti deleted the mp/add-segmentedcontrol-size-variant branch September 19, 2022 18:42
@primer-cssprimer-css mentioned this pull request Sep 19, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

SegmentedButton - cannot specify size of the button

2 participants

@mperrotti@colebemis