Skip to content

Fixes sx prop on the SegmentedControl buttons - #2355

Merged
mperrotti merged 2 commits into
mainfrom
mp/fix-segmented-control-sx-prop
Sep 20, 2022
Merged

Fixes sx prop on the SegmentedControl buttons#2355
mperrotti merged 2 commits into
mainfrom
mp/fix-segmented-control-sx-prop

Conversation

@mperrotti

Copy link
Copy Markdown
Contributor

Properly merges the sx prop when cloning button children.

Closes#2286

@mperrotti
mperrotti requested review from a team and siddharthkpSeptember 20, 2022 17:57
@changeset-bot

changeset-botBot commented Sep 20, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 129f006

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Sep 20, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js76.4 KB (+0.01% 🔺)
dist/browser.umd.js77.05 KB (+0.01% 🔺)

@mperrotti
mperrotti temporarily deployed to github-pages September 20, 2022 18:08 Inactive
@mperrotti
mperrotti merged commit e052644 into mainSep 20, 2022
@mperrotti
mperrotti deleted the mp/fix-segmented-control-sx-prop branch September 20, 2022 20:24
@primer-cssprimer-css mentioned this pull request Sep 20, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

SegmentedButton - sx props do not apply

2 participants

@mperrotti@colebemis