Skip to content

feat(TreeView): add label prop for Leading and Trailing Visual - #2434

Merged
joshblack merged 9 commits into
mainfrom
feat/add-labels-for-trailing-leading
Oct 17, 2022
Merged

feat(TreeView): add label prop for Leading and Trailing Visual#2434
joshblack merged 9 commits into
mainfrom
feat/add-labels-for-trailing-leading

Conversation

@joshblack

Copy link
Copy Markdown
Member

This PR adds in support for describing a TreeView.Item with aria-describedby and TreeView.LeadingItem and TreeView.TrailingItem. Each visual slot takes in an optional label prop which can be used to describe the overall TreeView.Item.

Background: https://github.com/github/primer/blob/main/apis/tree-view-api.md#leading-and-trailing-visuals

@joshblack
joshblack requested review from a team and rezrahOctober 14, 2022 21:01
@changeset-bot

changeset-botBot commented Oct 14, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 26d7dbe

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joshblack
joshblack requested review from colebemis and removed request for rezrahOctober 14, 2022 21:01
@github-actions

github-actionsBot commented Oct 14, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js77.92 KB (0%)
dist/browser.umd.js78.57 KB (0%)

Comment threadsrc/TreeView/TreeView.tsx
Comment thread.changeset/pretty-bats-cough.md Outdated
joshblackand others added 2 commits October 14, 2022 16:22
Co-authored-by: Cole Bemis <colebemis@github.com>
Comment threaddocs/content/TreeView.mdx Outdated

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great! Just left one question: #2434 (comment)

@joshblack
joshblack temporarily deployed to github-pages October 17, 2022 18:11 Inactive
@joshblack
joshblack merged commit 57c3b4d into mainOct 17, 2022
@joshblack
joshblack deleted the feat/add-labels-for-trailing-leading branch October 17, 2022 18:53
@primer-cssprimer-css mentioned this pull request Oct 17, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@colebemis