Skip to content

Move SplitPageLayout from drafts to alpha - #2435

Merged
colebemis merged 4 commits into
mainfrom
split-page-layout-alpha
Oct 14, 2022
Merged

Move SplitPageLayout from drafts to alpha#2435
colebemis merged 4 commits into
mainfrom
split-page-layout-alpha

Conversation

@colebemis

Copy link
Copy Markdown
Contributor

SplitPageLayout has been signed off by the accessibility team and is ready to be moved to alpha status. This PR updates the SplitPageLayout docs and exports SplitPageLayout from the main @primer/react bundle.

BeforeAfter
import {SplitPageLayout} from '@primer/react/draftsimport {SplitPageLayout} from '@primer/react'

@changeset-bot

changeset-botBot commented Oct 14, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: cf247c0

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@colebemis
colebemis marked this pull request as ready for review October 14, 2022 21:33
@colebemis
colebemis requested review from a team and pksjceOctober 14, 2022 21:33
@github-actions

github-actionsBot commented Oct 14, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js77.93 KB (+0.34% 🔺)
dist/browser.umd.js78.58 KB (+0.31% 🔺)

# description: TODO
source: https://github.com/primer/react/tree/main/src/PageLayout
storybook: https://primer.style/react/storybook?path=/story/layout-pagelayout--default
a11yReviewed: true

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

💖 thank you for remembering to add this in! I have a PR open to add a task in the issue template as a reminder.

@colebemis
colebemis temporarily deployed to github-pages October 14, 2022 21:42 Inactive
@colebemis
colebemis merged commit e335cc4 into mainOct 14, 2022
@colebemis
colebemis deleted the split-page-layout-alpha branch October 14, 2022 21:52
@primer-cssprimer-css mentioned this pull request Oct 14, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@colebemis@joshblack@lesliecdubs