Skip to content

TreeView: Use correct design tokens and icons - #2446

Merged
colebemis merged 6 commits into
mainfrom
treeview-tokens-icons
Oct 18, 2022
Merged

TreeView: Use correct design tokens and icons#2446
colebemis merged 6 commits into
mainfrom
treeview-tokens-icons

Conversation

@colebemis

@colebemiscolebemis commented Oct 17, 2022

Copy link
Copy Markdown
Contributor
  • Updates the color of TreeView.DirectoryIcon to use treeViewItem.directory.fill variable
  • Updates the color of the hover background on chevrons to use treeViewItem.chevron.hoverBg
  • Updates the chevron icon to the 12px version instead of 16px
BeforeAfter
CleanShot 2022-10-17 at 16 54 43@2xCleanShot 2022-10-17 at 16 52 22@2x

@changeset-bot

changeset-botBot commented Oct 17, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0646e39

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Oct 17, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js77.99 KB (+0.12% 🔺)
dist/browser.umd.js78.64 KB (+0.12% 🔺)

@colebemis
colebemis marked this pull request as ready for review October 17, 2022 23:55
@colebemis
colebemis requested a review from a teamOctober 17, 2022 23:55
@colebemis
colebemis temporarily deployed to github-pages October 18, 2022 00:01 Inactive
Comment threadpackage.json

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks good ✨

@colebemis
colebemis temporarily deployed to github-pages October 18, 2022 21:56 Inactive
@colebemis
colebemis merged commit e00d03c into mainOct 18, 2022
@colebemis
colebemis deleted the treeview-tokens-icons branch October 18, 2022 21:59
@primer-cssprimer-css mentioned this pull request Oct 18, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@colebemis@broccolinisoup