Skip to content

Align status component table to primer.style/status table - #2467

Merged
josepmartins merged 7 commits into
mainfrom
josepmartins/status-label-upates
Oct 26, 2022
Merged

Align status component table to primer.style/status table#2467
josepmartins merged 7 commits into
mainfrom
josepmartins/status-label-upates

Conversation

@josepmartins

Copy link
Copy Markdown
Contributor

Describe your changes here.

Align the accessbile column with the https://primer.style/status/ table

  • Use Label for empty accessible rows
  • Adjust label copies to 'Not reviewedandReviewed`
  • Add done functional colors

Screenshots

Please provide before/after screenshots for any visual changes

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@josepmartins
josepmartins requested review from a team and rezrahOctober 21, 2022 10:26
@changeset-bot

changeset-botBot commented Oct 21, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3825958

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@josepmartinsjosepmartins changed the title Josepmartins/status label upatesAlign status component table to primer.style/status tableOct 21, 2022
@josepmartins
josepmartins marked this pull request as draft October 21, 2022 10:28
@github-actions

github-actionsBot commented Oct 21, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js78.14 KB (0%)
dist/browser.umd.js78.79 KB (0%)

@josepmartins
josepmartinsforce-pushed the josepmartins/status-label-upates branch from d2abec9 to d76e26dCompareOctober 21, 2022 10:33
@josepmartins
josepmartins marked this pull request as ready for review October 21, 2022 10:34
@josepmartinsjosepmartins added the docs Documentation label Oct 21, 2022
@josepmartins
josepmartins temporarily deployed to github-pages October 21, 2022 10:42 Inactive

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🔥

@josepmartins
josepmartins temporarily deployed to github-pages October 23, 2022 16:04 Inactive
@josepmartins
josepmartins temporarily deployed to github-pages October 25, 2022 08:20 Inactive
@josepmartinsjosepmartins self-assigned this Oct 25, 2022
@josepmartins
josepmartins temporarily deployed to github-pages October 25, 2022 15:28 Inactive

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@josepmartins
josepmartins temporarily deployed to github-pages October 26, 2022 07:25 Inactive
@josepmartins
josepmartins merged commit d396c89 into mainOct 26, 2022
@josepmartins
josepmartins deleted the josepmartins/status-label-upates branch October 26, 2022 08:35
@primer-cssprimer-css mentioned this pull request Oct 26, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

docsDocumentation

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@josepmartins@broccolinisoup@joshblack