Skip to content

fix(test): update test-helpers to include unobserve for ResizeObserver - #2483

Merged
joshblack merged 4 commits into
mainfrom
2426-add-unobserver-to-resize-observer-mock
Nov 2, 2022
Merged

fix(test): update test-helpers to include unobserve for ResizeObserver#2483
joshblack merged 4 commits into
mainfrom
2426-add-unobserver-to-resize-observer-mock

Conversation

@joshblack

Copy link
Copy Markdown
Member

Closes #2426

Changelog

New

Changed

  • Add unobserve to ResizeObserver in test-helpers.tsx

Removed

@joshblack
joshblack requested review from a team and siddharthkp October 26, 2022 16:20
@changeset-bot

changeset-bot Bot commented Oct 26, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7a31663

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@primer/react Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actions Bot commented Oct 26, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

Path Size
dist/browser.esm.js 78.87 KB (0%)
dist/browser.umd.js 79.51 KB (0%)

@joshblack
joshblack temporarily deployed to github-pages October 26, 2022 16:26 Inactive
@joshblack

Copy link
Copy Markdown
Member Author

@siddharthkp let me know if you have any questions or if I'm missing anything from the original issue!

@siddharthkp siddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This is perfect, thank you!

@joshblack
joshblack enabled auto-merge (squash) November 2, 2022 16:15
@joshblack
joshblack merged commit 23647cf into main Nov 2, 2022
@joshblack
joshblack deleted the 2426-add-unobserver-to-resize-observer-mock branch November 2, 2022 16:23
@joshblack
joshblack temporarily deployed to github-pages November 2, 2022 16:24 Inactive
@primer-css primer-css mentioned this pull request Nov 2, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

[Bug] jest test helpers is missing ResizeObserver. unobserve

2 participants