Skip to content

fix(TreeView): add hcm styles for aria-current in TreeView - #2492

Merged
colebemis merged 4 commits into
mainfrom
fix/add-hcm-fix-for-aria-current
Nov 1, 2022
Merged

fix(TreeView): add hcm styles for aria-current in TreeView#2492
colebemis merged 4 commits into
mainfrom
fix/add-hcm-fix-for-aria-current

Conversation

@joshblack

@joshblackjoshblack commented Oct 27, 2022

Copy link
Copy Markdown
Member

This addresses the HCM feedback from: https://github.com/github/primer/issues/1114#issuecomment-1292865037

TreeView uses the HighlightText keyword for aria-current and focus styles

Screen.Recording.2022-10-31.at.3.38.13.PM.mov

@joshblack
joshblack requested review from a team and rezrahOctober 27, 2022 19:31
@changeset-bot

changeset-botBot commented Oct 27, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b154db4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joshblack

Copy link
Copy Markdown
MemberAuthor

@ericwbailey would love your feedback on what the best values would be for HCM here. Mainly have just been using keywords from: https://www.w3.org/TR/css-color-4/#css-system-colors or transparent outlines.

@github-actions

github-actionsBot commented Oct 27, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js78.14 KB (0%)
dist/browser.umd.js78.79 KB (0%)

@joshblack
joshblack temporarily deployed to github-pages October 27, 2022 19:39 Inactive
@joshblack
joshblack temporarily deployed to github-pages October 31, 2022 20:47 Inactive

@ericwbaileyericwbailey left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Heck yeah!

@colebemis
colebemis merged commit cbf0ccf into mainNov 1, 2022
@colebemis
colebemis deleted the fix/add-hcm-fix-for-aria-current branch November 1, 2022 16:39
@primer-cssprimer-css mentioned this pull request Nov 1, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@joshblack@ericwbailey@colebemis