Skip to content

UnderlineNav2: Remove focus zone and unnecessary 'Arrow keys' & 'Home' & 'End' button navigation support - #2550

Merged
broccolinisoup merged 4 commits into
mainfrom
broccolinisoup/UnderlineNav-sign-off-remediations-1
Nov 10, 2022
Merged

UnderlineNav2: Remove focus zone and unnecessary 'Arrow keys' & 'Home' & 'End' button navigation support#2550
broccolinisoup merged 4 commits into
mainfrom
broccolinisoup/UnderlineNav-sign-off-remediations-1

Conversation

@broccolinisoup

@broccolinisoupbroccolinisoup commented Nov 8, 2022

Copy link
Copy Markdown
Member

Addressing an accessibility sign-off review issue that described below.

The support for the arrow keys, and Home/End, to move through the items is unnecessary in a list of links, and should be removed.

The component seems to use a "roving tabindex" pattern, whereby one link has tabindex="0" and the others have tabindex="-1". As users are expected to tab through a list of links, this is completely inappropriate and could render content unreachable. All tabindex attributes must be removed.

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@changeset-bot

changeset-botBot commented Nov 8, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0ab1605

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Nov 8, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js78.89 KB (0%)
dist/browser.umd.js79.54 KB (0%)

@broccolinisoup
broccolinisoup temporarily deployed to github-pages November 8, 2022 03:40 Inactive
@broccolinisoup
broccolinisoupforce-pushed the broccolinisoup/UnderlineNav-sign-off-remediations-1 branch from 2c3943e to 3762e8cCompareNovember 8, 2022 03:51
@broccolinisoupbroccolinisoup changed the title UnderlineNav2: Remove 'Arrow keys' & 'Home' & 'End' button navigation supportUnderlineNav2: broccolinisoup/UnderlineNav-sign-off-remediations-1Nov 8, 2022
@broccolinisoup
broccolinisoup temporarily deployed to github-pages November 8, 2022 03:58 Inactive
@broccolinisoupbroccolinisoup changed the title UnderlineNav2: broccolinisoup/UnderlineNav-sign-off-remediations-1UnderlineNav2: Remove focus zone and unnecessary 'Arrow keys' & 'Home' & 'End' button navigation supportNov 8, 2022
@broccolinisoup
broccolinisoup marked this pull request as ready for review November 8, 2022 04:28
@broccolinisoup
broccolinisoup enabled auto-merge (squash) November 9, 2022 23:04
@broccolinisoup
broccolinisoup temporarily deployed to github-pages November 9, 2022 23:12 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages November 10, 2022 01:02 Inactive
@broccolinisoup
broccolinisoup merged commit e13e8ad into mainNov 10, 2022
@broccolinisoup
broccolinisoup deleted the broccolinisoup/UnderlineNav-sign-off-remediations-1 branch November 10, 2022 01:02
@primer-cssprimer-css mentioned this pull request Nov 10, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@broccolinisoup@ericwbailey@joshblack