Uh oh!
There was an error while loading. Please reload this page.
UnderlineNav2: Add aria-hidden and sr-only class support for a descriptive "More" button label - #2552
Merged
broccolinisoup merged 8 commits intoNov 10, 2022
Conversation
…een readers and add list for tul
🦋 Changeset detectedLatest commit: eaeff03 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Contributor
size-limit report 📦
|
broccolinisoup
marked this pull request as ready for review
November 8, 2022 04:39
broccolinisoup
requested review from
a team, colebemis, ericwbailey and joshblackNovember 8, 2022 04:39
ericwbailey
approved these changes
Nov 8, 2022
joshblack
requested changes
Nov 8, 2022
joshblack
left a comment
Member
There was a problem hiding this comment.
Left a comment above for a structure question, let me know what you think!
Uh oh!
There was an error while loading. Please reload this page.
…button label Co-authored-by: Josh Black <joshblack@github.com>
joshblack
approved these changes
Nov 9, 2022
broccolinisoup
enabled auto-merge (squash)
November 10, 2022 01:18
broccolinisoup
deleted the
broccolinisoup/UnderlineNav-sign-off-remediations-3
branch
November 10, 2022 01:38
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for freeto join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Addressing an accessibility sign-off review issue that described below.
Merge checklist
Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.