Skip to content

Update condensed mode for Safari workaround - #2569

Merged
joshblack merged 3 commits into
mainfrom
markdown-editor-resize-fix
Nov 16, 2022
Merged

Update condensed mode for Safari workaround#2569
joshblack merged 3 commits into
mainfrom
markdown-editor-resize-fix

Conversation

@nicolleromero

Copy link
Copy Markdown
Contributor

This PR provides a workaround for a bug in Safari that is preventing the MarkdownEditor from resizing (due to not being recalculated) when rendered in the condensed state.

This bug occurs after reopening the editor for the README a second (or subsequent) time and may be due to a bug introduced in version 15.5 of Safari. The bug is still present in Safari v16.0 and the proposed workaround does not affect behavior or rendering of the editor in Chrome or when not in condensed mode.

Fixeshttps://github.com/github/memex/issues/11141

Screenshots

*Both demos recorded in Safari v16.0 (also tested in Chrome)

BEFORE

MD.editor.resize.Safari.bug.mov

AFTER

MD.editor.resize.Safari.fix.mov

Non-condensed editor:

image

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@nicolleromero
nicolleromero requested review from a team and josepmartinsNovember 11, 2022 00:14
@changeset-bot

changeset-botBot commented Nov 11, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 8e714d9

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Nov 11, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js78.9 KB (0%)
dist/browser.umd.js79.54 KB (0%)

@nicolleromero
nicolleromero temporarily deployed to github-pages November 11, 2022 00:22 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2569 November 11, 2022 00:22 Inactive
@nicolleromero
nicolleromeroforce-pushed the markdown-editor-resize-fix branch from 136faac to 737e8abCompareNovember 14, 2022 00:43
@nicolleromero
nicolleromero temporarily deployed to github-pages November 14, 2022 00:50 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2569 November 14, 2022 00:50 Inactive
@nicolleromeronicolleromero changed the title Update condensed state for Safari workaroundUpdate condensed mode for Safari workaroundNov 14, 2022
@joshblackjoshblack added the 💓collab a vibrant hub of collaboration label Nov 14, 2022
@nicolleromero

Copy link
Copy Markdown
ContributorAuthor

@joshblack Thanks for the review! Are there any additional steps I should take for this fix (regarding the changeset and merging)?

@joshblack

Copy link
Copy Markdown
Member

@nicolleromero just pushed one up! I think with this it should be good to go 👍

@joshblack
joshblack enabled auto-merge (squash) November 16, 2022 17:13
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2569 November 16, 2022 17:17 Inactive
@joshblack
joshblack temporarily deployed to github-pages November 16, 2022 17:19 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-2569 November 16, 2022 17:19 Inactive
@joshblack
joshblack merged commit 0e7a638 into mainNov 16, 2022
@joshblack
joshblack deleted the markdown-editor-resize-fix branch November 16, 2022 17:35
@primer-cssprimer-css mentioned this pull request Nov 16, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

💓collaba vibrant hub of collaboration

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@nicolleromero@joshblack