Skip to content

Add useRefObjectAsForwardedRef to public hooks - #2591

Merged
iansan5653 merged 5 commits into
mainfrom
make-useRefObjectAsForwardedRef-public
Nov 18, 2022
Merged

Add useRefObjectAsForwardedRef to public hooks#2591
iansan5653 merged 5 commits into
mainfrom
make-useRefObjectAsForwardedRef-public

Conversation

@iansan5653

Copy link
Copy Markdown
Contributor

Adds the useRefObjectAsForwardedRef hook to the publicly exported hooks. This is a generic and useful hook that we can use in the Memex application and elsewhere.

@iansan5653
iansan5653 requested review from a team and joshblackNovember 17, 2022 16:53
@changeset-bot

changeset-botBot commented Nov 17, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7b5e507

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Nov 17, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js78.92 KB (+0.03% 🔺)
dist/browser.umd.js79.55 KB (+0.03% 🔺)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-2591 November 17, 2022 17:02 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2591 November 17, 2022 17:03 Inactive
@iansan5653
iansan5653 temporarily deployed to github-pages November 17, 2022 17:03 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2591 November 17, 2022 17:04 Inactive
@iansan5653
iansan5653 temporarily deployed to github-pages November 17, 2022 17:27 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-2591 November 17, 2022 17:28 Inactive

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🚀

@iansan5653
iansan5653 enabled auto-merge (squash) November 18, 2022 16:36
@iansan5653
iansan5653 temporarily deployed to github-pages November 18, 2022 16:42 Inactive
@iansan5653
iansan5653 merged commit 764bf6b into mainNov 18, 2022
@iansan5653
iansan5653 deleted the make-useRefObjectAsForwardedRef-public branch November 18, 2022 16:42
@joshblack

Copy link
Copy Markdown
Member

Trying to dig into usage of this in PRC, is this kind of acting like a useMergedRefs type hook or is there something distinct going on here with the useImperativeHandle?

It seems like this kind of acts like a superset in that it doesn't have to be an HTML element where hooks like a useMergedRefs would expect only HTML elements?

@primer-cssprimer-css mentioned this pull request Nov 18, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@iansan5653@joshblack@colebemis