Skip to content

Button alignment and size bug fixes - #2652

Merged
langermank merged 40 commits into
mainfrom
button-alignment-fixes
Dec 20, 2022
Merged

Button alignment and size bug fixes#2652
langermank merged 40 commits into
mainfrom
button-alignment-fixes

Conversation

@langermank

@langermanklangermank commented Dec 5, 2022

Copy link
Copy Markdown
Contributor

Note: These are the non breaking changes from this PR

Improvements

  • Add a block prop for full width
  • Add alignContent prop to align content to center or start
  • Use control sizing CSS variable values (not using CSS vars just yet)
  • Use height over padding for more control over sizing

Visual updates

Addresses a few things missing from this discussion: https://github.com/github/primer/discussions/1016

  • Slightly more condensed padding
  • Improved invisible hover states (will update these colors when new functional colors are addressed)

Closes:

Screenshots

Please refer to the comprehensive Storybook stories- this is just a small sample of some of the changes

beforeafterdescription
imageimagetrailing action slot alignment
imageimagedefault padding, font-weight
imageimagelarge height
icon was blueimageinvisible svg color
imageimagetrailing action locks right in full width
imageimageinvisible buttons use default text color if visuals are present, and accent blue if no visuals.

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@langermank
langermank requested review from a team and broccolinisoupDecember 5, 2022 22:01
@changeset-bot

changeset-botBot commented Dec 5, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: caf0f23

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Dec 5, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js84.91 KB (+0.46% 🔺)
dist/browser.umd.js85.52 KB (+0.42% 🔺)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 5, 2022 22:07 Inactive
@langermank
langermank temporarily deployed to github-pages December 5, 2022 22:09 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 5, 2022 22:10 Inactive
@langermank
langermank temporarily deployed to github-pages December 5, 2022 22:20 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 5, 2022 22:21 Inactive
@langermank
langermank temporarily deployed to github-pages December 5, 2022 22:27 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 5, 2022 22:27 Inactive
@langermanklangermank mentioned this pull request Dec 5, 2022
6 tasks
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 6, 2022 22:33 Inactive
@langermank
langermank temporarily deployed to github-pages December 6, 2022 22:37 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 6, 2022 22:38 Inactive
@langermank
langermank temporarily deployed to github-pages December 6, 2022 22:48 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 6, 2022 22:49 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 6, 2022
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 6, 2022
@primer
primerBot temporarily deployed to github-pages December 6, 2022 23:24 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 6, 2022 23:25 Inactive
Co-authored-by: Mike Perrotti <mperrotti@github.com>
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 19, 2022 22:23 Inactive
@langermank
langermank temporarily deployed to github-pages December 19, 2022 22:24 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 19, 2022 22:24 Inactive
@langermank
langermank temporarily deployed to github-pages December 19, 2022 22:51 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 19, 2022 22:51 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 20, 2022
@langermank
langermank temporarily deployed to github-pages December 20, 2022 00:28 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 20, 2022 00:28 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 20, 2022
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 20, 2022 00:41 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 20, 2022 00:45 Inactive
@langermank
langermank temporarily deployed to github-pages December 20, 2022 00:49 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 20, 2022 00:59 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 20, 2022 01:03 Inactive
@langermank
langermank temporarily deployed to github-pages December 20, 2022 02:31 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2652 December 20, 2022 02:32 Inactive
@langermank
langermank temporarily deployed to github-pages December 20, 2022 16:34 — with GitHub Actions Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-2652 December 20, 2022 16:35 Inactive
@primer-cssprimer-css mentioned this pull request Dec 20, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@langermank@broccolinisoup@mperrotti@joshblack