Skip to content

feat(project): update project to use React 18 - #2661

Merged
joshblack merged 37 commits into
mainfrom
ci/add-react-18-test
Dec 16, 2022
Merged

feat(project): update project to use React 18#2661
joshblack merged 37 commits into
mainfrom
ci/add-react-18-test

Conversation

@joshblack

@joshblackjoshblack commented Dec 6, 2022

Copy link
Copy Markdown
Member

Related: #2536

This PR is a rollup of the following PRs to update the project to use React 18 in development:

The strategy used is to test both React 17 and React 18. In the codebase, we also need to check for the presence of React 18 features before using them in order to maintain backwards compatability.

@changeset-bot

changeset-botBot commented Dec 6, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9f25f74

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Dec 6, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js84.08 KB (+0.06% 🔺)
dist/browser.umd.js84.73 KB (+0.09% 🔺)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-2661 December 6, 2022 23:32 Inactive
@joshblack
joshblack temporarily deployed to github-pages December 6, 2022 23:35 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2661 December 6, 2022 23:35 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2661 December 6, 2022 23:37 Inactive
@joshblackjoshblack added the skip changeset This change does not need a changelog label Dec 6, 2022
@joshblack
joshblack temporarily deployed to github-pages December 6, 2022 23:46 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2661 December 6, 2022 23:46 Inactive
@joshblack
joshblack marked this pull request as ready for review December 6, 2022 23:47

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks good to me, just left a question.

Comment threadscript/set-react-version.js
@joshblackjoshblack mentioned this pull request Dec 13, 2022
22 tasks
@joshblackjoshblack changed the title chore(project): add script to set react versionfeat(project): update project to use React 18Dec 15, 2022
@joshblack
joshblack temporarily deployed to github-pages December 16, 2022 18:10 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2661 December 16, 2022 18:10 Inactive
@joshblack
joshblack temporarily deployed to github-pages December 16, 2022 18:23 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2661 December 16, 2022 18:24 Inactive
@joshblackjoshblack reopened this Dec 16, 2022
@joshblackjoshblack removed the skip changeset This change does not need a changelog label Dec 16, 2022
@joshblackjoshblack reopened this Dec 16, 2022
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2661 December 16, 2022 19:13 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2661 December 16, 2022 19:14 Inactive
@joshblack
joshblack temporarily deployed to github-pages December 16, 2022 19:18 — with GitHub Actions Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-2661 December 16, 2022 19:18 Inactive
@joshblack
joshblack merged commit 5dd4bb1 into mainDec 16, 2022
@joshblack
joshblack deleted the ci/add-react-18-test branch December 16, 2022 19:41
@primer-cssprimer-css mentioned this pull request Dec 16, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@joshblack@broccolinisoup@colebemis