Skip to content

Add Noto Sans to fontStack - #2721

Merged
simurai merged 6 commits into
mainfrom
noto-sans
Dec 26, 2022
Merged

Add Noto Sans to fontStack#2721
simurai merged 6 commits into
mainfrom
noto-sans

Conversation

@simurai

@simuraisimurai commented Dec 22, 2022

Copy link
Copy Markdown
Contributor

This adds Noto Sans to the fontStack. It's a side-stream (not sure if that's a term 😆) from PCSS's primer/css#2300. It should improve alignment issues on Linux.

Closeshttps://github.com/github/primer/issues/1493

Screenshots

BeforeAfter
Screenshot from 2022-10-25 20-50-42Screenshot from 2022-10-25 20-58-03

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@simurai
simurai requested review from a team and josepmartinsDecember 22, 2022 07:26
@changeset-bot

changeset-botBot commented Dec 22, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 19a27c0

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Dec 22, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js84.93 KB (+0.02% 🔺)
dist/browser.umd.js85.53 KB (+0.02% 🔺)

@simurai
simurai temporarily deployed to github-pages December 22, 2022 07:34 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2721 December 22, 2022 07:34 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2721 December 22, 2022 07:34 Inactive

@josepmartinsjosepmartins left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

✨ 🚢 !

@colebemis
colebemis temporarily deployed to github-pages December 22, 2022 17:42 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2721 December 22, 2022 17:42 Inactive
@simurai
simurai enabled auto-merge (squash) December 23, 2022 08:34
@simurai
simurai temporarily deployed to github-pages December 23, 2022 08:43 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2721 December 23, 2022 08:43 Inactive
@simurai
simurai disabled auto-merge December 23, 2022 09:13
@simurai
simurai temporarily deployed to github-pages December 26, 2022 02:57 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2721 December 26, 2022 02:58 Inactive
@simurai
simurai temporarily deployed to github-pages December 26, 2022 03:08 — with GitHub Actions Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-2721 December 26, 2022 03:08 Inactive
@simurai
simurai merged commit 386561a into mainDec 26, 2022
@simurai
simurai deleted the noto-sans branch December 26, 2022 04:16
@primer-cssprimer-css mentioned this pull request Dec 26, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@simurai@josepmartins@colebemis