Uh oh!
There was an error while loading. Please reload this page.
Export new UnderlineNav in the main bundle and move the older one to the deprecated bundle - Major release - #2728
Export new UnderlineNav in the main bundle and move the older one to the deprecated bundle - Major release#2728broccolinisoup wants to merge 12 commits into
Conversation
🦋 Changeset detectedLatest commit: 96d86fc The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
size-limit report 📦
|
8772b4c to
882d9f3Compare882d9f3 to
01bcdaaComparecd5c184 to
e3f9bbfCompare5ead866 to
b383dbcCompare
Thanks @joshblack! I updated the branch and resolved all conflicts. The VRT and AAT checks are failing though and the |
joshblack
commented
Jan 27, 2023
Hey @broccolinisoup! 👋 I think when the id of the stories change (like removing drafts) the tests themselves will need to be changed. I pushed up a commit that I think should tackle the new paths 🤞 |
broccolinisoup
commented
Jan 27, 2023
Thanks so much @joshblack ! It is very interesting - I thought I did that step. I even documented it on the depreciation component docs but apparently I failed to apply that step on my own PR lol Thanks a bunch! |
Hi! This pull request has been marked as stale because it has been open with no activity for 60 days. You can comment on the pull request or remove the stale label to keep it open. If you do nothing, this pull request will be closed in 7 days. |
broccolinisoup
commented
May 14, 2023
Closing this in favor of #3260 |
Describe your changes here.
Closes # (type the issue number after # if applicable; otherwise remove this line)
Screenshots
Please provide before/after screenshots for any visual changes
Merge checklist
Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.