Skip to content

Remove shadow from MarkdownEditortextarea - #2765

Merged
joshblack merged 3 commits into
mainfrom
md-editor-shadow
Jan 11, 2023
Merged

Remove shadow from MarkdownEditortextarea#2765
joshblack merged 3 commits into
mainfrom
md-editor-shadow

Conversation

@iansan5653

Copy link
Copy Markdown
Contributor

There is a weird little subtle box-shadow on the MarkdownEditor input:

screenshot

This small fix removes that:

screenshot

@iansan5653
iansan5653 requested a review from a teamJanuary 11, 2023 17:43
@iansan5653iansan5653 self-assigned this Jan 11, 2023
@changeset-bot

changeset-botBot commented Jan 11, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 5215569

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Jan 11, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js87.38 KB (0%)
dist/browser.umd.js88.03 KB (0%)

@iansan5653
iansan5653 temporarily deployed to github-pages January 11, 2023 17:50 — with GitHub Actions Inactive

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🎉

@joshblackjoshblack added the 💓collab a vibrant hub of collaboration label Jan 11, 2023
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2765 January 11, 2023 17:50 Inactive
@joshblack
joshblack enabled auto-merge (squash) January 11, 2023 17:50
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2765 January 11, 2023 17:50 Inactive
@joshblack
joshblack temporarily deployed to github-pages January 11, 2023 17:56 — with GitHub Actions Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-2765 January 11, 2023 17:57 Inactive
@joshblack
joshblack merged commit 154db4b into mainJan 11, 2023
@joshblack
joshblack deleted the md-editor-shadow branch January 11, 2023 17:59
@primer-cssprimer-css mentioned this pull request Jan 11, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

💓collaba vibrant hub of collaboration

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@iansan5653@joshblack