Skip to content

1573 accessibility request primer react link enforce what as can be set to with a runtime validation rule 1572 - #2789

Merged
agreenberry merged 25 commits into
mainfrom
1573-accessibility-request-primer-react-link-enforce-what-as-can-be-set-to-with-a-runtime-validation-rule-1572
Jan 27, 2023
Merged

1573 accessibility request primer react link enforce what as can be set to with a runtime validation rule 1572#2789
agreenberry merged 25 commits into
mainfrom
1573-accessibility-request-primer-react-link-enforce-what-as-can-be-set-to-with-a-runtime-validation-rule-1572

Conversation

@agreenberry

@agreenberryagreenberry commented Jan 20, 2023

Copy link
Copy Markdown
Contributor

Describe your changes here.

Closes#1573 (type the issue number after # if applicable; otherwise remove this line)

The user sees a warning if they implement Link without using an a or button element in the as prop.

Screenshots

Please provide before/after screenshots for any visual changes
image

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@changeset-bot

changeset-botBot commented Jan 20, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 44453f8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Jan 20, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js88.95 KB (+0.1% 🔺)
dist/browser.umd.js89.56 KB (+0.1% 🔺)

@agreenberry
agreenberry temporarily deployed to github-pages January 20, 2023 18:36 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 20, 2023 18:36 Inactive
Co-authored-by: Cole Bemis <colebemis@users.noreply.github.com>
@agreenberry
agreenberryforce-pushed the 1573-accessibility-request-primer-react-link-enforce-what-as-can-be-set-to-with-a-runtime-validation-rule-1572 branch from d6c9fd4 to 77aaa4fCompareJanuary 20, 2023 19:20
@agreenberry
agreenberry temporarily deployed to github-pages January 20, 2023 19:27 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 20, 2023 19:27 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 23, 2023 23:08 Inactive
@agreenberry
agreenberry temporarily deployed to github-pages January 23, 2023 23:09 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 23, 2023 23:10 Inactive
@agreenberry
agreenberry marked this pull request as ready for review January 23, 2023 23:22
@agreenberry
agreenberry requested review from a team and colebemisJanuary 23, 2023 23:22
@agreenberry
agreenberry temporarily deployed to github-pages January 23, 2023 23:26 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 23, 2023 23:26 Inactive
…-enforce-what-as-can-be-set-to-with-a-runtime-validation-rule-1572
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 24, 2023 16:49 Inactive
@agreenberry
agreenberry temporarily deployed to github-pages January 24, 2023 16:49 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 24, 2023 16:50 Inactive
@agreenberry
agreenberry enabled auto-merge (squash) January 24, 2023 17:55
Comment threadsrc/Link.tsx Outdated
Co-authored-by: Cole Bemis <colebemis@github.com>
@agreenberry
agreenberry temporarily deployed to github-pages January 24, 2023 22:10 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 24, 2023 22:10 Inactive
…-enforce-what-as-can-be-set-to-with-a-runtime-validation-rule-1572
@agreenberry
agreenberry temporarily deployed to github-pages January 25, 2023 18:02 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 25, 2023 18:03 Inactive
agreenberryand others added 2 commits January 27, 2023 11:27
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 27, 2023 16:35 Inactive
@agreenberry
agreenberry temporarily deployed to github-pages January 27, 2023 16:36 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 27, 2023 16:37 Inactive
@agreenberry
agreenberry temporarily deployed to github-pages January 27, 2023 17:31 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 27, 2023 17:31 Inactive
agreenberryand others added 3 commits January 27, 2023 13:39
…-enforce-what-as-can-be-set-to-with-a-runtime-validation-rule-1572
…at-as-can-be-set-to-with-a-runtime-validation-rule-1572' of github.com:primer/react into 1573-accessibility-request-primer-react-link-enforce-what-as-can-be-set-to-with-a-runtime-validation-rule-1572
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2789 January 27, 2023 18:45 Inactive
@agreenberry
agreenberry temporarily deployed to github-pages January 27, 2023 18:49 — with GitHub Actions Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-2789 January 27, 2023 18:49 Inactive
@agreenberry
agreenberry merged commit 44f66cb into mainJan 27, 2023
@agreenberry
agreenberry deleted the 1573-accessibility-request-primer-react-link-enforce-what-as-can-be-set-to-with-a-runtime-validation-rule-1572 branch January 27, 2023 19:41
@primer-cssprimer-css mentioned this pull request Jan 27, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@agreenberry@colebemis