Skip to content

PageHeader: A11y eng review remediations - #2900

Merged
broccolinisoup merged 9 commits into
mainfrom
a11y-page-header
Feb 16, 2023
Merged

PageHeader: A11y eng review remediations#2900
broccolinisoup merged 9 commits into
mainfrom
a11y-page-header

Conversation

@broccolinisoup

@broccolinisoupbroccolinisoup commented Feb 13, 2023

Copy link
Copy Markdown
Member

Describe your changes here.

This PR addresses issues found and implement recommendations from PageHeaderaccessibility eng review . Please see this comment for more info on each item below.

Issues addressed:

  • Remove default value of aria-label on ParentLink component.

Recommendations implemented:

  • Allow usage of as=”nav | div” on PageHeader.Navigation component.

Also Added a feature story to cover custom components that can be passed to PageHeader.Navigation as children

Screenshots

No visual changes.

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@changeset-bot

changeset-botBot commented Feb 13, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2347578

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Feb 13, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js89.74 KB (0%)
dist/browser.umd.js90.28 KB (0%)

@broccolinisoup
broccolinisoup temporarily deployed to github-pages February 13, 2023 05:59 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2900 February 13, 2023 06:00 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages February 13, 2023 06:46 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2900 February 13, 2023 06:46 Inactive
@broccolinisoupbroccolinisoup changed the title A11y page headerPageHeader: A11y eng review remediationsFeb 13, 2023
@broccolinisoup
broccolinisoup temporarily deployed to github-pages February 13, 2023 06:54 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2900 February 13, 2023 06:54 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages February 13, 2023 08:00 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2900 February 13, 2023 08:00 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2900 February 13, 2023 08:10 Inactive
@primer
primerBot temporarily deployed to github-pages February 13, 2023 08:11 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2900 February 13, 2023 08:12 Inactive
@broccolinisoup
broccolinisoup marked this pull request as ready for review February 13, 2023 09:02

@TylerJDevTylerJDev left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great! Love the additional tests added too. 🎉

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Left a comment for wrapping the warning in a __DEV__ check, should be good to go after that!

Comment threadsrc/PageHeader/PageHeader.tsx Outdated
@broccolinisoup
broccolinisoup temporarily deployed to github-pages February 15, 2023 05:26 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2900 February 15, 2023 05:26 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages February 15, 2023 23:34 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2900 February 15, 2023 23:35 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages February 16, 2023 04:37 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2900 February 16, 2023 04:37 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages February 16, 2023 05:31 — with GitHub Actions Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-2900 February 16, 2023 05:32 Inactive
@broccolinisoup
broccolinisoup added this pull request to the merge queueFeb 16, 2023
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Feb 16, 2023
@broccolinisoup
broccolinisoup added this pull request to the merge queueFeb 16, 2023
Merged via the queue into main with commit 3768cd7Feb 16, 2023
@broccolinisoup
broccolinisoup deleted the a11y-page-header branch February 16, 2023 09:17
@primer-cssprimer-css mentioned this pull request Feb 16, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@broccolinisoup@joshblack@TylerJDev