Skip to content

Promote TreeView to beta 🎉 - #2903

Merged
colebemis merged 8 commits into
mainfrom
treeview-sign-off
Feb 16, 2023
Merged

Promote TreeView to beta 🎉#2903
colebemis merged 8 commits into
mainfrom
treeview-sign-off

Conversation

@colebemis

@colebemiscolebemis commented Feb 16, 2023

Copy link
Copy Markdown
Contributor

TreeView has been signed-off by @jscholes and now meets all the criteria for the beta status.

04C14CB4-7C14-4B88-8C3E-54B61742F098.MOV

Video courtesy of @jonrohan

Changelog

Updated

  • The docs reflect TreeView's a11y sign-off and beta status
  • TreeView is now exported from the main bundle

@colebemis
colebemis requested review from a team and langermankFebruary 16, 2023 03:08
@changeset-bot

changeset-botBot commented Feb 16, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e296b63

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Feb 16, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js94.5 KB (+5.31% 🔺)
dist/browser.umd.js95.08 KB (+5.32% 🔺)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-2903 February 16, 2023 03:15 Inactive
@colebemis
colebemis temporarily deployed to github-pages February 16, 2023 03:16 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2903 February 16, 2023 03:17 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2903 February 16, 2023 03:18 Inactive

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🎉

@colebemis
colebemis temporarily deployed to github-pages February 16, 2023 04:50 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2903 February 16, 2023 04:51 Inactive
@colebemiscolebemis changed the title Promote TreeView to betaPromote TreeView to beta 🎉Feb 16, 2023
@colebemis
colebemis temporarily deployed to github-pages February 16, 2023 17:05 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2903 February 16, 2023 17:05 Inactive
@colebemis
colebemis added this pull request to the merge queueFeb 16, 2023
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Feb 16, 2023
@colebemis
colebemis temporarily deployed to github-pages February 16, 2023 17:38 — with GitHub Actions Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-2903 February 16, 2023 17:38 Inactive
@colebemis
colebemis added this pull request to the merge queueFeb 16, 2023
Merged via the queue into main with commit 13651baFeb 16, 2023
@colebemis
colebemis deleted the treeview-sign-off branch February 16, 2023 17:55
@primer-cssprimer-css mentioned this pull request Feb 16, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@colebemis@broccolinisoup