Skip to content

Add story source code to components.json - #2948

Merged
colebemis merged 18 commits into
mainfrom
story-source-code
Feb 28, 2023
Merged

Add story source code to components.json#2948
colebemis merged 18 commits into
mainfrom
story-source-code

Conversation

@colebemis

@colebemiscolebemis commented Feb 28, 2023

Copy link
Copy Markdown
Contributor

This PR adds source code for storybook stories to the generated components.json file so that we can display that source code in the storybook embeds on primer.style/design (example: https://primer.style/design/components/tree-view/react#examples)

CleanShot 2023-02-27 at 19 26 01@2x

Before/After

generated/component.json:

CleanShot 2023-02-27 at 19 27 59@2x

@changeset-bot

changeset-botBot commented Feb 28, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c4e9346

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Feb 28, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js94.49 KB (0%)
dist/browser.umd.js95.07 KB (0%)

@colebemis
colebemis marked this pull request as ready for review February 28, 2023 02:30
@colebemis
colebemis requested review from a team, josepmartins and joshblackFebruary 28, 2023 02:30
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2948 February 28, 2023 02:31 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2948 February 28, 2023 02:36 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2948 February 28, 2023 02:37 Inactive
@primer
primerBot temporarily deployed to github-pages February 28, 2023 02:39 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2948 February 28, 2023 02:40 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2948 February 28, 2023 02:40 Inactive
@colebemiscolebemis changed the title Add story source code to component.jsonAdd story source code to components.jsonFeb 28, 2023
@primer
primerBot temporarily deployed to github-pages February 28, 2023 03:14 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2948 February 28, 2023 03:14 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2948 February 28, 2023 03:15 Inactive

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This is such an exciting improvement! 🚀

Comment threadscript/components-json/build.ts

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Woohoo! 🥳 Just left a comment for the leading ; but looks amazing 🔥

@github-actions
github-actionsBot temporarily deployed to storybook-preview-2948 February 28, 2023 16:52 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2948 February 28, 2023 16:53 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2948 February 28, 2023 16:53 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2948 February 28, 2023 16:58 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-2948 February 28, 2023 17:00 Inactive
@colebemis
colebemis added this pull request to the merge queueFeb 28, 2023
Merged via the queue into main with commit 0215c96Feb 28, 2023
@colebemis
colebemis deleted the story-source-code branch February 28, 2023 17:21
@primer-cssprimer-css mentioned this pull request Feb 28, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@colebemis@joshblack@langermank