Skip to content

Hidden: rename prop on -> when and reformat stories and add e2e tests - #2985

Merged
broccolinisoup merged 7 commits into
mainfrom
refactor-hidden
Mar 8, 2023
Merged

Hidden: rename prop on -> when and reformat stories and add e2e tests#2985
broccolinisoup merged 7 commits into
mainfrom
refactor-hidden

Conversation

@broccolinisoup

@broccolinisoupbroccolinisoup commented Mar 6, 2023

Copy link
Copy Markdown
Member

Renames the prop from being on -> when as discussed with the team on React Eng Sync. Because Hidden is a draft component, it is okay to release this as minor.

Also reformats the storybook and addes e2e tests.

For reformatting, the tracking issue: https://github.com/github/primer/issues/1488

Add storybook stories for Hidden

Checklist

  • Ensure your chosen component has a component directory in the src directory
  • Ensure the component has a root story file in the component directory
  • Ensure that the root story file exports a Default story
  • (Optional) Create a Playground story
  • Move all other stories for the component into a .features.stories.tsx file
  • Opt the component into storybook tests
  • (Optional) Create visual regression tests for the component

@broccolinisoupbroccolinisoup added the skip changeset This change does not need a changelog label Mar 6, 2023
@changeset-bot

changeset-botBot commented Mar 6, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e9309cf

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Mar 6, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js94.68 KB (0%)
dist/browser.umd.js95.25 KB (0%)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-2985 March 6, 2023 06:18 Inactive
@primer
primerBot temporarily deployed to github-pages March 6, 2023 06:19 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2985 March 6, 2023 06:19 Inactive
@broccolinisoupbroccolinisoup added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Mar 6, 2023
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2985 March 8, 2023 02:55 Inactive
@primer
primerBot temporarily deployed to github-pages March 8, 2023 02:56 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2985 March 8, 2023 02:57 Inactive
@broccolinisoupbroccolinisoup added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 8, 2023
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 8, 2023
@primer
primerBot temporarily deployed to github-pages March 8, 2023 03:36 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2985 March 8, 2023 03:37 Inactive
@broccolinisoup
broccolinisoup marked this pull request as ready for review March 8, 2023 04:05
@broccolinisoup
broccolinisoup requested review from a team and langermankMarch 8, 2023 04:05
@broccolinisoup
broccolinisoup added this pull request to the merge queueMar 8, 2023
Merged via the queue into main with commit 38ac17aMar 8, 2023
@broccolinisoup
broccolinisoup deleted the refactor-hidden branch March 8, 2023 20:32
@primer-cssprimer-css mentioned this pull request Mar 8, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

skip changesetThis change does not need a changelog

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@broccolinisoup@joshblack