Skip to content

Make StateLabel's 'status' prop required - #2993

Merged
mperrotti merged 8 commits into
mainfrom
mp/statelabel-status-rqd
Mar 15, 2023
Merged

Make StateLabel's 'status' prop required#2993
mperrotti merged 8 commits into
mainfrom
mp/statelabel-status-rqd

Conversation

@mperrotti

Copy link
Copy Markdown
Contributor

Closes#2955

Screenshots

Please provide before/after screenshots for any visual changes

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@mperrotti
mperrotti requested review from a team and josepmartinsMarch 6, 2023 22:54
@changeset-bot

changeset-botBot commented Mar 6, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b9ffe2f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Mar 6, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js94.52 KB (0%)
dist/browser.umd.js95.1 KB (0%)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-2993 March 6, 2023 23:01 Inactive
@primer
primerBot temporarily deployed to github-pages March 6, 2023 23:02 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2993 March 6, 2023 23:03 Inactive

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks good! I think with the type tests it's fair to remove the shouldAcceptCallWithNoProps and should be good to go after that 👍

@mperrotti
mperrotti temporarily deployed to github-pages March 14, 2023 17:23 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2993 March 14, 2023 17:24 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2993 March 15, 2023 21:07 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-2993 March 15, 2023 21:10 Inactive
@mperrotti
mperrotti added this pull request to the merge queueMar 15, 2023
Merged via the queue into main with commit 6aa9205Mar 15, 2023
@mperrotti
mperrotti deleted the mp/statelabel-status-rqd branch March 15, 2023 21:22
@primer-cssprimer-css mentioned this pull request Mar 15, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

StateLabel does not render if the status prop is undefined

2 participants

@mperrotti@joshblack