Skip to content

Tooltip: Update the component markup for a11y remediations & refactor the existing styles towards more static styling - #3032

Merged
broccolinisoup merged 38 commits into
mainfrom
tooltip-experiments
Apr 11, 2023
Merged

Tooltip: Update the component markup for a11y remediations & refactor the existing styles towards more static styling#3032
broccolinisoup merged 38 commits into
mainfrom
tooltip-experiments

Conversation

@broccolinisoup

@broccolinisoupbroccolinisoup commented Mar 15, 2023

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/1909
Closeshttps://github.com/github/primer/issues/1907
Closeshttps://github.com/github/primer/issues/1918

Storybook features: https://primer-7bad896c1f-13348165.drafts.github.io/storybook/?path=/story/components-tooltip-features--tooltip-description-type-tooltip
Docs: https://primer-7bad896c1f-13348165.drafts.github.io/Tooltip

This PR refactors the current Tooltip component to update the markup and styles to match with the expected accessibility requirements.

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@broccolinisoupbroccolinisoup added the skip changeset This change does not need a changelog label Mar 15, 2023
@changeset-bot

changeset-botBot commented Mar 15, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7a816fb

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Mar 15, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js95.82 KB (+0.33% 🔺)
dist/browser.umd.js96.45 KB (+0.42% 🔺)

@broccolinisoup
broccolinisoup temporarily deployed to github-pages March 15, 2023 04:15 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3032 March 15, 2023 04:16 Inactive
@broccolinisoupbroccolinisoup changed the title Tooltip: Refactor the markup without styled components and apply a11y remediationsWIP - Tooltip: Refactor the markup without styled components and apply a11y remediationsMar 15, 2023
@broccolinisoup
broccolinisoup temporarily deployed to github-pages March 16, 2023 06:57 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3032 March 16, 2023 06:58 Inactive
Comment threadsrc/Tooltip2.tsx Outdated
Comment threadsrc/Tooltip2.tsx Outdated
Comment threadsrc/Tooltip2.tsx Outdated
@broccolinisoup
broccolinisoup temporarily deployed to github-pages March 16, 2023 08:22 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3032 March 16, 2023 08:23 Inactive
@broccolinisoupbroccolinisoup changed the title WIP - Tooltip: Refactor the markup without styled components and apply a11y remediationsTooltip: Refactor the markup without styled components and apply a11y remediationsMar 20, 2023
@broccolinisoup
broccolinisoup marked this pull request as ready for review March 20, 2023 02:59
@broccolinisoup
broccolinisoup requested a review from a teamMarch 20, 2023 02:59
@broccolinisoup
broccolinisoup temporarily deployed to github-pages March 20, 2023 03:04 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3032 March 20, 2023 03:04 Inactive
Comment threadsrc/Tooltip2.tsx Outdated
Comment threadsrc/Tooltip2.tsx Outdated
@broccolinisoup
broccolinisoup temporarily deployed to github-pages March 20, 2023 09:46 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3032 April 5, 2023 05:20 Inactive
@broccolinisoup

Copy link
Copy Markdown
MemberAuthor

Note that the snapshots are failing on main. It should be good once we resolve the issue on main

@TylerJDevTylerJDev left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Amazing work @broccolinisoup! 🤩

This tooltip will be a huge improvement over our current one in terms of accessibility. As far as my comments, you've addressed all of them! Thanks so much for all the work on this! 🎉

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🥳

@broccolinisoup
broccolinisoup temporarily deployed to github-pages April 5, 2023 21:46 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3032 April 5, 2023 21:46 Inactive
Comment threadsrc/Tooltip/Tooltip.docs.json
Comment threadsrc/Tooltip/Tooltip.docs.json
@broccolinisoup
broccolinisoup temporarily deployed to github-pages April 10, 2023 20:23 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3032 April 10, 2023 20:23 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3032 April 11, 2023 00:50 Inactive
@primer
primerBot temporarily deployed to github-pages April 11, 2023 00:50 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3032 April 11, 2023 00:51 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages April 11, 2023 01:04 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3032 April 11, 2023 01:04 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages April 11, 2023 01:30 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3032 April 11, 2023 01:30 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages April 11, 2023 01:49 — with GitHub Actions Inactive
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants

@broccolinisoup@TylerJDev@joshblack@langermank@edersonlucas