Skip to content

Export useResponsiveValue in index.ts - #3090

Merged
kendallgassner merged 3 commits into
mainfrom
export-useResponsiveValue
Mar 31, 2023
Merged

Export useResponsiveValue in index.ts#3090
kendallgassner merged 3 commits into
mainfrom
export-useResponsiveValue

Conversation

@kendallgassner

Copy link
Copy Markdown
Contributor

Describe your changes here.

Exposing useResponsiveValue in index.ts

Closes # (type the issue number after # if applicable; otherwise remove this line)

Screenshots

Please provide before/after screenshots for any visual changes

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@kendallgassner
kendallgassner requested review from a team and colebemisMarch 29, 2023 22:22
@changeset-bot

changeset-botBot commented Mar 29, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 77c6136

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Mar 29, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js95.05 KB (+0.02% 🔺)
dist/browser.umd.js95.59 KB (+0.02% 🔺)

@kendallgassner
kendallgassner temporarily deployed to github-pages March 29, 2023 22:29 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3090 March 29, 2023 22:29 Inactive

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

👍 Looks like you just need to update the snapshot test (npm test -- -u)

@kendallgassner
kendallgassner temporarily deployed to github-pages March 31, 2023 20:36 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3090 March 31, 2023 20:37 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3090 March 31, 2023 20:47 Inactive
@kendallgassner
kendallgassner added this pull request to the merge queueMar 31, 2023
Merged via the queue into main with commit 1b8bb16Mar 31, 2023
@kendallgassner
kendallgassner deleted the export-useResponsiveValue branch March 31, 2023 21:04
@primer-cssprimer-css mentioned this pull request Mar 31, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@kendallgassner@colebemis