Skip to content

Bump @primer/primitives to 7.11.3 - #3093

Merged
simurai merged 10 commits into
mainfrom
primitives@7.11.3
Apr 4, 2023
Merged

Bump @primer/primitives to 7.11.3#3093
simurai merged 10 commits into
mainfrom
primitives@7.11.3

Conversation

@simurai

@simuraisimurai commented Mar 30, 2023

Copy link
Copy Markdown
Contributor

This bumps @primer/primitives to

Screenshots

The snapshot diff is quite larger since the default text color changed. The most noticeable difference is the Primary button:

BeforeAfter
Screen Shot 2023-03-29 at 19 28 02Screen Shot 2023-03-29 at 19 25 52

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@changeset-bot

changeset-botBot commented Mar 30, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 91dcc79

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Mar 30, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js95.15 KB (-0.04% 🔽)
dist/browser.umd.js95.72 KB (-0.03% 🔽)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3093 March 30, 2023 02:42 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3093 March 30, 2023 02:46 Inactive
@primer
primerBot temporarily deployed to github-pages March 30, 2023 02:47 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3093 March 30, 2023 02:48 Inactive
```
npm run build
npm run test -- -u
```
@simurai
simurai temporarily deployed to github-pages March 30, 2023 03:29 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3093 March 30, 2023 03:30 Inactive
```
npm test -- -u
```
@simurai
simurai temporarily deployed to github-pages March 30, 2023 04:17 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3093 March 30, 2023 04:17 Inactive
@simuraisimurai added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 30, 2023
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 30, 2023
@primer
primerBot temporarily deployed to github-pages March 30, 2023 04:40 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3093 March 30, 2023 04:40 Inactive
@simurai
simurai temporarily deployed to github-pages March 30, 2023 08:44 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3093 March 30, 2023 08:45 Inactive
@simurai
simurai temporarily deployed to github-pages March 30, 2023 09:36 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3093 March 30, 2023 09:36 Inactive
@simurai
simurai temporarily deployed to github-pages March 30, 2023 10:00 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3093 March 30, 2023 10:01 Inactive
@simurai
simurai marked this pull request as ready for review March 30, 2023 11:01
@simurai
simurai requested review from a team and mperrottiMarch 30, 2023 11:01
@simurai
simurai enabled auto-merge April 4, 2023 00:42
@simurai
simurai disabled auto-merge April 4, 2023 00:42
@simurai
simurai enabled auto-merge April 4, 2023 00:53
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3093 April 4, 2023 00:57 Inactive
@simurai
simurai added this pull request to the merge queueApr 4, 2023
Merged via the queue into main with commit 8b5307dApr 4, 2023
@simurai
simurai deleted the primitives@7.11.3 branch April 4, 2023 01:13
@primer-cssprimer-css mentioned this pull request Apr 4, 2023
@simuraisimurai mentioned this pull request Apr 5, 2023
6 tasks
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@simurai@lukasoppermann@joshblack