Skip to content

UnderlineNav2: Add transparent outline for focus ring to support WHCM - #3165

Merged
broccolinisoup merged 3 commits into
mainfrom
bs/underlineNav-whcm
Apr 13, 2023
Merged

UnderlineNav2: Add transparent outline for focus ring to support WHCM#3165
broccolinisoup merged 3 commits into
mainfrom
bs/underlineNav-whcm

Conversation

@broccolinisoup

Copy link
Copy Markdown
Member

It came up on PageHeader accessibility sign-off review that the UnderlineNav links' focus rings are not visible on the Window High Contrast Theme. Added a transparent outline as suggested.

Closeshttps://github.com/github/primer/issues/2081 (Hubbers only link)

Screenshots

Before

Screenshot 2023-04-12 at 2 32 44 pm

After

Screenshot 2023-04-13 at 3 40 31 pm

Merge checklist

  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@broccolinisoup
broccolinisoup requested review from a team and langermankApril 13, 2023 06:01
@changeset-bot

changeset-botBot commented Apr 13, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f7ca3dd

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Apr 13, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js95.78 KB (+0.01% 🔺)
dist/browser.umd.js96.39 KB (+0.02% 🔺)

@broccolinisoup
broccolinisoup temporarily deployed to github-pages April 13, 2023 06:08 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3165 April 13, 2023 06:08 Inactive
@broccolinisoup
broccolinisoup added this pull request to the merge queueApr 13, 2023
Merged via the queue into main with commit d749f49Apr 13, 2023
@broccolinisoup
broccolinisoup deleted the bs/underlineNav-whcm branch April 13, 2023 23:18
This was referenced Apr 13, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@broccolinisoup@joshblack