Skip to content

font weight bold for ActionList.Item with inline description - #3170

Merged
lukasoppermann merged 4 commits into
mainfrom
ActionList-with-desc
Apr 14, 2023
Merged

font weight bold for ActionList.Item with inline description#3170
lukasoppermann merged 4 commits into
mainfrom
ActionList-with-desc

Conversation

@lukasoppermann

@lukasoppermannlukasoppermann commented Apr 14, 2023

Copy link
Copy Markdown
Contributor

When an inline description is added to the item the normal text gets bold. This is how it is designed and specified in the docs: https://primer.style/design/components/action-list

Screenshots

BeforeAfter
Screenshot 2023-04-14 at 08 50 43Screenshot 2023-04-14 at 08 50 51

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@changeset-bot

changeset-botBot commented Apr 14, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ff6c5a2

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Apr 14, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js95.88 KB (+0.02% 🔺)
dist/browser.umd.js96.48 KB (+0.02% 🔺)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3170 April 14, 2023 06:58 Inactive
@lukasoppermannlukasoppermann self-assigned this Apr 14, 2023
@lukasoppermann
lukasoppermann temporarily deployed to github-pages April 14, 2023 07:01 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3170 April 14, 2023 07:01 Inactive
Comment threadsrc/ActionList/Item.tsx Outdated
Co-authored-by: Cole Bemis <colebemis@github.com>
@lukasoppermann
lukasoppermann temporarily deployed to github-pages April 14, 2023 07:32 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3170 April 14, 2023 07:32 Inactive
@lukasoppermannlukasoppermann added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 14, 2023
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 14, 2023
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3170 April 14, 2023 07:55 Inactive
@lukasoppermann
lukasoppermann added this pull request to the merge queueApr 14, 2023
Merged via the queue into main with commit 1f2df9bApr 14, 2023
@lukasoppermann
lukasoppermann deleted the ActionList-with-desc branch April 14, 2023 08:06
This was referenced Apr 14, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@lukasoppermann@colebemis