Skip to content

Use SSR-compatible slot implementation in MarkdownEditor - #3174

Merged
colebemis merged 2 commits into
mainfrom
markdowneditor-slots
Apr 20, 2023
Merged

Use SSR-compatible slot implementation in MarkdownEditor#3174
colebemis merged 2 commits into
mainfrom
markdowneditor-slots

Conversation

@colebemis

Copy link
Copy Markdown
Contributor

Swaps out the underlying slots implementation for MarkdownEditor.

See #1690 for more context.

@colebemis
colebemis requested review from a team and mperrottiApril 14, 2023 22:35
@changeset-bot

changeset-botBot commented Apr 14, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7c4affb

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Apr 14, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js95.88 KB (0%)
dist/browser.umd.js96.48 KB (0%)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3174 April 14, 2023 22:42 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3174 April 14, 2023 22:46 Inactive
@colebemis
colebemis added this pull request to the merge queueApr 20, 2023
Merged via the queue into main with commit d4c6ae5Apr 20, 2023
@colebemis
colebemis deleted the markdowneditor-slots branch April 20, 2023 22:56
@primer-cssprimer-css mentioned this pull request Apr 20, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@colebemis@mperrotti