Skip to content

3096 remove aria live=polite on textinput wrapper - #3182

Merged
agreenberry merged 12 commits into
mainfrom
3096-remove-aria-live=polite-on-textinput-wrapper
Apr 27, 2023
Merged

3096 remove aria live=polite on textinput wrapper#3182
agreenberry merged 12 commits into
mainfrom
3096-remove-aria-live=polite-on-textinput-wrapper

Conversation

@agreenberry

@agreenberryagreenberry commented Apr 17, 2023

Copy link
Copy Markdown
Contributor

Describe your changes here.

Closes#3096

Screenshots

Please provide before/after screenshots for any visual changes
image
image

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@agreenberry
agreenberry requested a review from a teamApril 17, 2023 15:01
@agreenberryagreenberry linked an issue Apr 17, 2023 that may be closed by this pull request
@changeset-bot

changeset-botBot commented Apr 17, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 8e6ef01

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Apr 17, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js121.18 KB (0%)
dist/browser.umd.js121.47 KB (+0.01% 🔺)

@agreenberry
agreenberry temporarily deployed to github-pages April 17, 2023 15:07 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3182 April 17, 2023 15:07 Inactive
@agreenberry
agreenberry temporarily deployed to github-pages April 20, 2023 18:37 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3182 April 20, 2023 18:38 Inactive
auto-merge was automatically disabled April 24, 2023 22:34

Pull request was closed

@agreenberry
agreenberryforce-pushed the 3096-remove-aria-live=polite-on-textinput-wrapper branch from 9fe09da to cb233cfCompareApril 24, 2023 22:34
@agreenberry
agreenberry temporarily deployed to github-pages April 24, 2023 22:56 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3182 April 24, 2023 22:56 Inactive
@agreenberry
agreenberry temporarily deployed to github-pages April 25, 2023 15:37 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3182 April 25, 2023 15:38 Inactive
@agreenberry
agreenberry temporarily deployed to github-pages April 25, 2023 16:46 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3182 April 25, 2023 16:46 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3182 April 25, 2023 17:18 Inactive
@agreenberry
agreenberry temporarily deployed to github-pages April 25, 2023 17:19 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3182 April 25, 2023 17:20 Inactive
@agreenberry
agreenberry temporarily deployed to github-pages April 25, 2023 17:47 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3182 April 25, 2023 17:47 Inactive
@agreenberry
agreenberry temporarily deployed to github-pages April 25, 2023 19:07 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3182 April 25, 2023 19:07 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3182 April 27, 2023 16:38 Inactive
@agreenberry
agreenberry added this pull request to the merge queueApr 27, 2023
Merged via the queue into main with commit 7827c71Apr 27, 2023
@agreenberry
agreenberry deleted the 3096-remove-aria-live=polite-on-textinput-wrapper branch April 27, 2023 16:57
This was referenced Apr 27, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Remove aria-live="polite" on TextInput wrapper

3 participants

@agreenberry@colebemis@jonrohan