Skip to content

Underlinenav2: Add a z-index on the disclosure menu - #3200

Merged
broccolinisoup merged 3 commits into
mainfrom
bs/underline-stacking-fix
May 2, 2023
Merged

Underlinenav2: Add a z-index on the disclosure menu#3200
broccolinisoup merged 3 commits into
mainfrom
bs/underline-stacking-fix

Conversation

@broccolinisoup

@broccolinisoupbroccolinisoup commented Apr 20, 2023

Copy link
Copy Markdown
Member

Describe your changes here.

UnderlineNav2 disclosure menu has a stacking issue and when there are other components they overlaps the menu. Adding a z-index to make sure it appears on the top of the stack.

Note: The menu was intentionally is built with the native HTML rather than using ActionMenu as it is a portal and being a portal was making the component inaccessible.

Closeshttps://github.com/github/primer/issues/1943 (Hubbers only link)

Screenshots

Before

Screenshot 2023-04-20 at 4 37 11 pm

After

Screenshot 2023-04-20 at 5 05 06 pm

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@changeset-bot

changeset-botBot commented Apr 20, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d635c9a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Apr 20, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js95.37 KB (-0.01% 🔽)
dist/browser.umd.js95.91 KB (-0.01% 🔽)

@broccolinisoup
broccolinisoup temporarily deployed to github-pages April 20, 2023 07:01 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3200 April 20, 2023 07:02 Inactive
@broccolinisoup
broccolinisoup marked this pull request as ready for review April 20, 2023 07:11
@broccolinisoup
broccolinisoup temporarily deployed to github-pages April 20, 2023 07:18 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3200 April 20, 2023 07:18 Inactive

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Oh z-index 😅

@broccolinisoup
broccolinisoup added this pull request to the merge queueMay 2, 2023
Merged via the queue into main with commit 63d8ad6May 2, 2023
@broccolinisoup
broccolinisoup deleted the bs/underline-stacking-fix branch May 2, 2023 03:30
@primer-cssprimer-css mentioned this pull request May 2, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@broccolinisoup@joshblack@colebemis