Skip to content

Remove --primer prefix from CSS var - #3209

Merged
langermank merged 4 commits into
mainfrom
css-var-update
Apr 21, 2023
Merged

Remove --primer prefix from CSS var#3209
langermank merged 4 commits into
mainfrom
css-var-update

Conversation

@langermank

Copy link
Copy Markdown
Contributor

Preparing for Primitives v8. This will be bumped in dotcom soon, and since it has a fallback there should be no issues renaming it right now.

If there are more CSS vars being used that I couldn't find, let me know!

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

@langermank
langermank requested review from a team and colebemisApril 21, 2023 18:35
@changeset-bot

changeset-botBot commented Apr 21, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c145ae0

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Apr 21, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js95.37 KB (-0.01% 🔽)
dist/browser.umd.js95.91 KB (-0.01% 🔽)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3209 April 21, 2023 18:41 Inactive
@langermank
langermank temporarily deployed to github-pages April 21, 2023 18:43 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3209 April 21, 2023 18:43 Inactive
@langermank
langermank added this pull request to the merge queueApr 21, 2023
Merged via the queue into main with commit a627953Apr 21, 2023
@langermank
langermank deleted the css-var-update branch April 21, 2023 19:20
@primer-cssprimer-css mentioned this pull request Apr 21, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@colebemis