Skip to content

feat(DataTable): add Table.ErrorDialog component - #3276

Merged
joshblack merged 7 commits into
mainfrom
feat/add-error-dialog-component
May 19, 2023
Merged

feat(DataTable): add Table.ErrorDialog component#3276
joshblack merged 7 commits into
mainfrom
feat/add-error-dialog-component

Conversation

@joshblack

Copy link
Copy Markdown
Member

This PR adds in an ErrorDialog component to mirror TreeView.ErrorDialog but for the DataTable set of components.

Changelog

New

  • Add Table.ErrorDialog component
  • Add "With Network Error" story

Changed

Removed

@joshblack
joshblack requested review from a team and colebemisMay 11, 2023 17:04
@changeset-bot

changeset-botBot commented May 11, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 241fcc4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joshblackjoshblack changed the title feat(DataTable): add ErrorDialog componentfeat(DataTable): add Table.ErrorDialog componentMay 11, 2023
@github-actions

github-actionsBot commented May 11, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js115.97 KB (0%)
dist/browser.umd.js117.81 KB (0%)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3276 May 11, 2023 17:11 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3276 May 11, 2023 17:12 Inactive
@primer
primerBot temporarily deployed to github-pages May 11, 2023 17:14 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3276 May 11, 2023 17:14 Inactive

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great!

const id = url.toString()
if (!cache.has(id)) {
cache.set(id, Repo.pageInfo(pageSize))
export function useFlakeyQuery<T>({

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

haha love it ❤️

@joshblack
joshblack enabled auto-merge May 15, 2023 19:28
@joshblack
joshblack temporarily deployed to github-pages May 15, 2023 19:35 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3276 May 15, 2023 19:35 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3276 May 19, 2023 14:43 Inactive
@joshblack
joshblack added this pull request to the merge queueMay 19, 2023
Merged via the queue into main with commit 8abf268May 19, 2023
@joshblack
joshblack deleted the feat/add-error-dialog-component branch May 19, 2023 15:00
This was referenced May 19, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@colebemis