Skip to content

CounterLabel should forward ref and dom props - #3291

Merged
mattcosta7 merged 3 commits into
mainfrom
counter-label-should-forward-refs
May 18, 2023
Merged

CounterLabel should forward ref and dom props#3291
mattcosta7 merged 3 commits into
mainfrom
counter-label-should-forward-refs

Conversation

@mattcosta7

@mattcosta7mattcosta7 commented May 16, 2023

Copy link
Copy Markdown
Contributor

CounterLabel should forward both dom props and refs tot he underlying box

Closes # (type the issue number after # if applicable; otherwise remove this line)

Screenshots

Please provide before/after screenshots for any visual changes

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@mattcosta7
mattcosta7 requested review from a team and siddharthkpMay 16, 2023 20:39
@changeset-bot

changeset-botBot commented May 16, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e7cbc86

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mattcosta7mattcosta7 changed the title add passthrough props to coutnerlabelCounterLabel should forward ref and dom propsMay 16, 2023
@mattcosta7mattcosta7 self-assigned this May 16, 2023
@github-actions

github-actionsBot commented May 16, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js115.96 KB (+0.01% 🔺)
dist/browser.umd.js117.79 KB (+0.01% 🔺)

@mattcosta7
mattcosta7 temporarily deployed to github-pages May 16, 2023 20:45 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3291 May 16, 2023 20:46 Inactive
sx,
)}
{...props}
as="span"

Copy link
Copy Markdown
ContributorAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

we don't pass as through (in ts land) so avoiding it getting passed through by non-ts code

@mattcosta7
mattcosta7force-pushed the counter-label-should-forward-refs branch from fbb5a61 to 6468085CompareMay 16, 2023 20:51
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3291 May 16, 2023 20:53 Inactive
@mattcosta7
mattcosta7 temporarily deployed to github-pages May 16, 2023 20:58 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3291 May 16, 2023 20:59 Inactive

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great! Just had a style question and wanted to hear what your thoughts were 👀

@mattcosta7

Copy link
Copy Markdown
ContributorAuthor

Looks great! Just had a style question and wanted to hear what your thoughts were 👀

I don't see the question!

@mattcosta7
mattcosta7 added this pull request to the merge queueMay 18, 2023
Merged via the queue into main with commit 1378bc1May 18, 2023
@mattcosta7
mattcosta7 deleted the counter-label-should-forward-refs branch May 18, 2023 04:01
This was referenced May 18, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@mattcosta7@joshblack